)]}'
{"id":"openstack%2Fskyline-console~894609","triplet_id":"openstack%2Fskyline-console~master~Ie63f8d330ad5442fdb5c6263e83e974e53fb1a3e","project":"openstack/skyline-console","branch":"master","topic":"instance","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-09-12 07:06:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie63f8d330ad5442fdb5c6263e83e974e53fb1a3e","subject":"feat: add detail info to action in detail tab","status":"MERGED","created":"2023-09-12 03:16:46.000000000","updated":"2023-09-12 07:07:33.000000000","submitted":"2023-09-12 07:06:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894609-instance","meta_rev_id":"288fb7feda8b3a4b9883b852c914464ce9cf8498","_number":894609,"virtual_id_number":894609,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-12 07:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-09-12 05:34:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-09-12 05:34:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-12 04:47:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-12 05:34:37.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"572c13f63ab64c67cbc5955cb135174404bbcbf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-12 03:16:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ca9f0c0245051d944fecbbdfef86c724c2a4b5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 04:47:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f3b04b30d3c446d85446d68c43d9a21\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d00b55d004f347e9ae78aa318fe73996 : SUCCESS in 4m 47s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/e3362b26d8284bd5b497f490b516cc0b : SUCCESS in 11m 06s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e019b307ffdf459fa501c5fc37cc348e : SUCCESS in 7m 47s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/fe66c8a0dfae41ea929018bb219a0f57 : SUCCESS in 7m 25s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8c9e76c57a72437dac956223509320e9 : SUCCESS in 1h 25m 31s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5e9b9dbdb3704e0895825c00aa00d5c2 : SUCCESS in 1h 16m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2af9a149aa78ca0d6efcc2160cc390cdbfb4c4ed","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-09-12 05:34:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d960ecf58dc814d0c35310cdd6720272725f7647","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 05:35:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a52412e8e9acb542680ac0895e46ceaa44cb168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 07:06:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5b1cfd71ba14704be87044fed23694d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c37aac7ea3c24ea88f3fadbc5de2e749 : SUCCESS in 4m 21s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0ca97842582546d79c0a44f729582924 : SUCCESS in 7m 20s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/3cb13107e33941b9ab7de7da5261b14c : SUCCESS in 5m 50s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d26dab2204c54fbc83fe4722a1244971 : SUCCESS in 5m 42s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/85048b0df7b542d38fb91c4ce53a1065 : SUCCESS in 1h 26m 19s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/451cd3db40704834a4348badb29165e3 : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d184901f5dd573fd45216cb29e88b1b19e922547","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 07:06:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"288fb7feda8b3a4b9883b852c914464ce9cf8498","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 07:07:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e72b5ea929e344b8be742c531ae59d44\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c9a82a38dc6478e99e6ea7f948a0f5f : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61e07535a5edeb457f4c47a4e9fb669c93255a80","revisions":{"61e07535a5edeb457f4c47a4e9fb669c93255a80":{"kind":"REWORK","_number":1,"created":"2023-09-12 03:16:46.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/09/894609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/894609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/894609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/894609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/894609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/894609/1"}}},"commit":{"parents":[{"commit":"d5b58ac16dfe20b6173893c92937aa102abdee40","subject":"feat: add action props to ModalAction component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d5b58ac16dfe20b6173893c92937aa102abdee40"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-12 03:16:20.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-12 03:16:20.000000000","tz":480},"subject":"feat: add detail info to action in detail tab","message":"feat: add detail info to action in detail tab\n\n1. add instance detail info to the attach volume action in the instance detail page\n2. add instance detail info to the manage security group action and detach action in the instance detail page\n3. add port detail info to the manage security group action and detach action in the port detail page\n\nChange-Id: Ie63f8d330ad5442fdb5c6263e83e974e53fb1a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/61e07535a5edeb457f4c47a4e9fb669c93255a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/61e07535a5edeb457f4c47a4e9fb669c93255a80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
