)]}'
{"id":"openstack%2Fskyline-console~894916","triplet_id":"openstack%2Fskyline-console~master~I27904475edc48404271bea1c42634f23be8ffbd1","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"34680":{"account":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"last_update":"2023-11-27 03:36:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I27904475edc48404271bea1c42634f23be8ffbd1","subject":"Edit the Object Storage \u003e Public Access directory listing link","status":"MERGED","created":"2023-09-13 13:04:26.000000000","updated":"2023-11-27 03:37:17.000000000","submitted":"2023-11-27 03:36:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894916","meta_rev_id":"cafcaf2f59c037f58c3610f8b85f8ecc76aced93","_number":894916,"virtual_id_number":894916,"owner":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-27 03:36:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-11-27 01:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2023-11-27 01:34:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-11-27 01:33:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-13 14:29:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-27 01:22:27.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2023-11-27 01:34:15.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"b823e38e5ba9e71e8661d01bc55c960715f1ab48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-09-13 13:04:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b900223cdd610baf438c7d1ca314914a1e5f66bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 14:29:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e68d5faacb64065bcbe44b48b9197f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a80b7657106445cae99b1ef43954a54 : SUCCESS in 5m 03s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/5a1e497ed7d348a8998c635f5234b878 : SUCCESS in 10m 50s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/3bfef990ee2c42458c81636707de38cc : SUCCESS in 7m 22s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/748a6117677246599e75ef3064bbcbc2 : SUCCESS in 7m 49s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/212d4e0547d7414cbdb1a0591c68f762 : FAILURE in 1h 18m 07s","accounts_in_message":[],"_revision_number":1},{"id":"aa4ad703742f826c7a6867c42a1ff36fd1dca013","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-11-24 11:54:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb55e2c73847188056b6a737895819ef6a65fbc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 12:38:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6350ad6a449e499b85a050306a522035\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fe74a4bcbad4d458c8e651bb9fa35ba : SUCCESS in 4m 41s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3207ab77a8a44f349f1ee6810d36e85b : SUCCESS in 6m 57s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4e0e9ccfb4e14fc1ab24cd13d323df19 : SUCCESS in 5m 14s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/7778ae0322d54c8d924f1bbc8b3cf634 : SUCCESS in 8m 03s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f66ca7cf22a648ec9a06ff690e4182d3 : SUCCESS in 42m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1bd07daa50f75a55f259a6631980041d2d03ff04","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-27 01:22:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d145af3c84f40c5b99a2e9e13ce0ec265100abe7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 01:22:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb1b1537e8b4b1eb6c7dbea737ed00512577c76a","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-27 01:25:10.000000000","message":"Patch Set 1: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"86aa5d76e35958bdc905b014c1f173bf83b71444","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 01:25:22.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"bd2b2e17967adbe8b1a253d5d7e02679a51feb84","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-27 01:33:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6e0b7d2f3ae083fccf33f8c98f70e990ed12126","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-27 01:34:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f511d58c8a810ff905b07e701f5b90f9522ac1b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 02:36:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41d91a8b034249d5b9ee3170aba933f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde3c6174d304574a531672d75822237 : SUCCESS in 4m 07s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/e87966e328d742e0b4d3f40a10ac65be : SUCCESS in 7m 16s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/27b80e1943864583b0e3b4d754d8708a : SUCCESS in 6m 56s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/193a6c50bd194aeb9bc33d52bc115b5d : SUCCESS in 6m 07s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b0c0b0070bb5467aa3242683a5a023ee : SUCCESS in 57m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6c6c0ffd838e74973ca29982386e72f871bf1304","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 02:37:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff6d91a2d25ef75e36883473ccd05a403add6444","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 03:36:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a967d061cff48af9cc24f2778c0ff6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1277449eb9eb4c35b7b770cbc4205efc : SUCCESS in 5m 26s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/51ae45288e754e52bf419306b04377aa : SUCCESS in 12m 24s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c405ad2ffc6a45238fad0373f290f457 : SUCCESS in 7m 37s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/9f4d0f1ded3e45049aa6097a38c43c92 : SUCCESS in 7m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ef7daf362c3a42d2b811387c87d6d5a0 : SUCCESS in 53m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9b1f6fa65d5a769ec8b23c990eb3550f2aa00afe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 03:36:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cafcaf2f59c037f58c3610f8b85f8ecc76aced93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 03:37:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23acef25c7fb4a7a8de925f2316997fc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/034c08c0c83942d4be47fac3117d2c03 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"312c27c038c61b18bfe65b2775f03eb4335e44bd","revisions":{"312c27c038c61b18bfe65b2775f03eb4335e44bd":{"kind":"REWORK","_number":1,"created":"2023-09-13 13:04:26.000000000","uploader":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"ref":"refs/changes/16/894916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/16/894916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/894916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/894916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/894916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/16/894916/1"}}},"commit":{"parents":[{"commit":"0f25664254f28f5f697ed6da7a97ce1cf1998bd9","subject":"Merge \"admin-menu.jsx updated.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0f25664254f28f5f697ed6da7a97ce1cf1998bd9"}]}],"author":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-09-13 12:55:36.000000000","tz":0},"committer":{"name":"Numan MENEKŞE","email":"numan.menekse@tubitak.gov.tr","date":"2023-09-13 13:03:18.000000000","tz":0},"subject":"Edit the Object Storage \u003e Public Access directory listing link","message":"Edit the Object Storage \u003e Public Access directory listing link\n\nOn the Storage \u003e Object Storage page, the directory listing page was not working because the \"Public Access\" link in the Detail Info popup was redirecting.\nAdded \"/\" to the end of the link address.\n\nChange-Id: I27904475edc48404271bea1c42634f23be8ffbd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/312c27c038c61b18bfe65b2775f03eb4335e44bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/312c27c038c61b18bfe65b2775f03eb4335e44bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
