)]}'
{"id":"openstack%2Fskyline-console~895674","triplet_id":"openstack%2Fskyline-console~stable%2F2023.2~Ie3831b3d575c156245d99762892a9a19ba2e727f","project":"openstack/skyline-console","branch":"stable/2023.2","topic":"bugs-stable/2023.2","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-09-19 09:08:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie3831b3d575c156245d99762892a9a19ba2e727f","subject":"fix: fix i18n for affinity(not mandatory) server group","status":"MERGED","created":"2023-09-19 08:16:53.000000000","updated":"2023-09-19 09:09:53.000000000","submitted":"2023-09-19 09:08:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895674-bugs-stable/2023.2","cherry_pick_of_change":895768,"cherry_pick_of_patch_set":1,"meta_rev_id":"6c9b0791c157a8e5894103ff6cab6519428e3f6a","_number":895674,"virtual_id_number":895674,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-19 09:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-09-19 08:51:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-09-19 08:51:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-19 08:33:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-19 08:51:46.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"5c005883ca1d8cac2ec609560179db18912e5700","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-19 08:16:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6a4fb7a4d9f48c1560d74b6dd809b8695dd2fe53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 08:33:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de87a31219674b4099bde5c76d6675ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/049aa48336cf4687bc2a193e1ab1ce22 : SUCCESS in 4m 33s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3958776b83a94f8ab3ae10c19b58f2c1 : SUCCESS in 10m 50s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/3c58782d0bb64c5fb6d31c7c2ee4e76f : SUCCESS in 5m 48s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/8ed99ef4ece84d86b4ff3fcf1e3a03eb : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":1},{"id":"adefcc5b6ce96d5c6c3428979b55b70402d1647b","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-09-19 08:51:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"afc0e0f2654ff3d198ab6dfb6daa9fd2889252be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 08:52:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6b73f98a11d14876ba7be35bc0a3c1330dcbd21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 09:08:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f908aef66ea4d0f8990bb9a4dcd5816\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db1b7170a42441ad981156a6d04fde14 : SUCCESS in 4m 50s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/894f748a788a48fdb21a2465a3fc7443 : SUCCESS in 10m 15s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/2f69843593da4fe6b6f56c96f522fbc6 : SUCCESS in 7m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/72f0ad0213804a52ac77e2c17d405c20 : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":1},{"id":"eeadc6340a89ebe18bba0ee073e8c30f62a0a162","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 09:08:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c9b0791c157a8e5894103ff6cab6519428e3f6a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 09:09:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d0973d8651c45a796c4089ca1623e5d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7201f4a539f846678431ea0da7567d04 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"524a0888d9a2a9f94be47e1594560704f6370a21","revisions":{"524a0888d9a2a9f94be47e1594560704f6370a21":{"kind":"REWORK","_number":1,"created":"2023-09-19 08:16:53.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/74/895674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/74/895674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/895674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/895674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/895674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/74/895674/1"}}},"commit":{"parents":[{"commit":"0f25664254f28f5f697ed6da7a97ce1cf1998bd9","subject":"Merge \"admin-menu.jsx updated.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0f25664254f28f5f697ed6da7a97ce1cf1998bd9"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-19 07:44:58.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-19 08:16:53.000000000","tz":0},"subject":"fix: fix i18n for affinity(not mandatory) server group","message":"fix: fix i18n for affinity(not mandatory) server group\n\nfix the tip of the affinity(not mandatory) server group\n\nChange-Id: Ie3831b3d575c156245d99762892a9a19ba2e727f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/524a0888d9a2a9f94be47e1594560704f6370a21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/524a0888d9a2a9f94be47e1594560704f6370a21"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
