)]}'
{"id":"openstack%2Fskyline-console~895768","triplet_id":"openstack%2Fskyline-console~master~Ie3831b3d575c156245d99762892a9a19ba2e727f","project":"openstack/skyline-console","branch":"master","topic":"bugs","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-09-19 08:27:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie3831b3d575c156245d99762892a9a19ba2e727f","subject":"fix: fix i18n for affinity(not mandatory) server group","status":"MERGED","created":"2023-09-19 07:46:01.000000000","updated":"2023-09-19 08:28:26.000000000","submitted":"2023-09-19 08:27:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895768-bugs","meta_rev_id":"73408e0d990fd6b10dc8f2da29d90514bff52fd8","_number":895768,"virtual_id_number":895768,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-19 08:27:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-09-19 08:12:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-09-19 08:12:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-19 08:03:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-19 08:12:35.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"518c134ae9558d74d2239cae4d5f31cdf2366ac6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-19 07:46:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47f1d43890202857d55d5479ab45b04dd751c96a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 08:03:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de4d222152bf46f28d10a7bd9591c496\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417e70e24db74fda94fd8c8c9bf6ba63 : SUCCESS in 4m 40s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/1a62c0a1e8474d92a3bb210ac28bcb34 : SUCCESS in 10m 16s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/5c79fc68b40b4d4d8e2ba3ab2ccf6928 : SUCCESS in 7m 10s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/7c0eb5e027ae49e5a44f7b0318bb2812 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"64928d3e3ae4cd6c6d4af25510e0cb5bf0c35df1","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-09-19 08:12:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"89be735e5b45e0ed4730ffa75d8a1ae3f58ac80e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 08:12:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"553c845eabb52ee63679c18d564a3170aca658cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 08:27:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bee1a06e546d41aca941bb29a8b20f5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f05d546099674baebbef7659b2569402 : SUCCESS in 5m 03s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/49b28c04f6e545bf93ee4f693ba88c24 : SUCCESS in 6m 39s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b5e3b5c5916e4e54b595fb5621f3383a : SUCCESS in 7m 00s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/c11d26fb162f4c41ba50c59369a6df8d : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3f5b8b7a334fa57d8edaf3641321c621815d4b4d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 08:27:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"73408e0d990fd6b10dc8f2da29d90514bff52fd8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 08:28:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616a43cbd94d4d8d8286ac6f876f7c49\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89c64192627d4448a7ba5c3a24521542 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc07a3506de1c8f19ee913b3dd65fd769938d076","revisions":{"dc07a3506de1c8f19ee913b3dd65fd769938d076":{"kind":"REWORK","_number":1,"created":"2023-09-19 07:46:01.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/68/895768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/68/895768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/895768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/895768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/68/895768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/68/895768/1"}}},"commit":{"parents":[{"commit":"0f25664254f28f5f697ed6da7a97ce1cf1998bd9","subject":"Merge \"admin-menu.jsx updated.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0f25664254f28f5f697ed6da7a97ce1cf1998bd9"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-19 07:44:58.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-19 07:44:58.000000000","tz":480},"subject":"fix: fix i18n for affinity(not mandatory) server group","message":"fix: fix i18n for affinity(not mandatory) server group\n\nfix the tip of the affinity(not mandatory) server group\n\nChange-Id: Ie3831b3d575c156245d99762892a9a19ba2e727f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dc07a3506de1c8f19ee913b3dd65fd769938d076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dc07a3506de1c8f19ee913b3dd65fd769938d076"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
