)]}'
{"id":"openstack%2Fskyline-console~896526","triplet_id":"openstack%2Fskyline-console~stable%2F2023.2~I4e12b423be7662ba83f8efbf55e1b6aa0e31105d","project":"openstack/skyline-console","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-09-27 07:59:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e12b423be7662ba83f8efbf55e1b6aa0e31105d","subject":"fix: show project id and filter data by it","status":"MERGED","created":"2023-09-27 06:32:59.000000000","updated":"2023-09-27 08:00:45.000000000","submitted":"2023-09-27 07:59:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896526","cherry_pick_of_change":896375,"cherry_pick_of_patch_set":4,"meta_rev_id":"2f4be596b8a80682548ddeb872ac5fbd221ba69b","_number":896526,"virtual_id_number":896526,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-27 07:59:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-09-27 07:43:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-09-27 07:43:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-27 06:44:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-27 07:43:35.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"5221b6ad7c43fe435ab3049ba98b025fc5bfe51c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-09-27 06:32:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ae54c3320c8372c00dc8f8a97530c7f8353a61f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 06:44:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5ced16858774393a547b4a8594a0faa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e4d2646b6e46318341d0f71046acab : SUCCESS in 4m 31s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/2933a56a25654f13b8048bad3ba9e187 : SUCCESS in 6m 29s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c807ca7d21274baf8b3915597ceba122 : SUCCESS in 5m 18s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/75016b3e513f432c8f7507d8cdcfecdf : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3dc9579d5ecb6822fd95aa29f66f2eda0a8f9808","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-09-27 07:43:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca3afe76343eb7a94360796b40311f7bf23bd923","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 07:43:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4bf583262dd0683a489bf44e4f62bda50d999f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 07:59:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/456b2bc4e97941898ad855d80bd06f16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/708576542298448baf5cd462b9b7c864 : SUCCESS in 5m 27s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/1bac457efe7c424ba768be34135d7f31 : SUCCESS in 11m 02s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6c0dec08550f4a5dbb878d5d6dfd0d88 : SUCCESS in 7m 41s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e5fd99823cf44eb19071a5f75904fd68 : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6ec3fca642efe0c566005bce8b7990d843a1f83b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 07:59:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2f4be596b8a80682548ddeb872ac5fbd221ba69b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 08:00:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5a32ab3a59c46b880edee62df44aa85\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32af9812712944c8869a17bd5033d00c : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3985526495892914ce8af20b06dd3a042eb8a02a","revisions":{"3985526495892914ce8af20b06dd3a042eb8a02a":{"kind":"REWORK","_number":1,"created":"2023-09-27 06:32:59.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/26/896526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/26/896526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/896526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/896526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/26/896526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/26/896526/1"}}},"commit":{"parents":[{"commit":"668c3e22c3e157cfdb299abb6cd936b2c0b9dc0f","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2023.2\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/668c3e22c3e157cfdb299abb6cd936b2c0b9dc0f"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-09-25 10:27:22.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-09-27 06:32:59.000000000","tz":0},"subject":"fix: show project id and filter data by it","message":"fix: show project id and filter data by it\n\nShow project id and filter data by it when create nework,subnet,fip,qos in management platform\n\nCloses-Bug: #2037275\nChange-Id: I4e12b423be7662ba83f8efbf55e1b6aa0e31105d\n(cherry picked from commit ec21bd17ab25ae4f668d77e5ef38d39b5cafe137)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3985526495892914ce8af20b06dd3a042eb8a02a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3985526495892914ce8af20b06dd3a042eb8a02a"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
