)]}'
{"id":"openstack%2Fskyline-console~896709","triplet_id":"openstack%2Fskyline-console~stable%2F2023.2~I42502b86b5361f129f4304b94ff294d3ea15c43f","project":"openstack/skyline-console","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-09-28 02:42:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I42502b86b5361f129f4304b94ff294d3ea15c43f","subject":"fix: remove content validator about certification","status":"MERGED","created":"2023-09-28 01:39:16.000000000","updated":"2023-09-28 02:44:25.000000000","submitted":"2023-09-28 02:42:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896709","cherry_pick_of_change":896615,"cherry_pick_of_patch_set":2,"meta_rev_id":"a714dd4fbf47d474cf824a4daa1cfbebc858fc63","_number":896709,"virtual_id_number":896709,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-28 02:42:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-09-28 02:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-09-28 02:23:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 01:55:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-28 02:23:39.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"26bfefcdea395fd2f4bbe96e19dfa2fdc196e378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-09-28 01:39:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c245b3c6c1ada6ab736d4f9f345c512f06aef1b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 01:55:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61c0b683836845ad8e2286936c4abf5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc8ef7532e743a88c87e9533350f688 : SUCCESS in 4m 46s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c63e0b032e1042f7b1ac1241390ddeb5 : SUCCESS in 11m 09s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/22a1551f8a564bfca5ece28e4a80847b : SUCCESS in 8m 00s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/872a3ce2a48e4b2c8c4ef52daeaf1d02 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6615c4296f998d86d74f1d2aa4322d493e147192","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-09-28 02:23:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08ff03552eab76b7f1b6dffdbb71cb57eb858340","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 02:24:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a03af1b4f3695c709656da826a1effe95421b01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 02:42:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14d5ecffba08408088ee36ce8df8e23c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7c332d15b0e421c85aed4e9237bfb23 : SUCCESS in 6m 50s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/1abd590cd194419eaf3681b3dee50da7 : SUCCESS in 11m 28s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7683404387ad4bedb6694b1a336edf3d : SUCCESS in 9m 47s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/c3f5f90d233f431480bb6f250b612161 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":1},{"id":"59036d63187c876cd8ed3ee5ef5386d280afc629","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 02:42:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a714dd4fbf47d474cf824a4daa1cfbebc858fc63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 02:44:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e402db7293a0468e829d78597fc22fdd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efac9f8d81884c09844180ed488a06e7 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a516dfd8f940ea03610167b12a75f413d402a60","revisions":{"7a516dfd8f940ea03610167b12a75f413d402a60":{"kind":"REWORK","_number":1,"created":"2023-09-28 01:39:16.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/09/896709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/896709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/896709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/896709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/896709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/896709/1"}}},"commit":{"parents":[{"commit":"3985526495892914ce8af20b06dd3a042eb8a02a","subject":"fix: show project id and filter data by it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3985526495892914ce8af20b06dd3a042eb8a02a"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-09-27 08:09:08.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-09-28 01:39:16.000000000","tz":0},"subject":"fix: remove content validator about certification","message":"fix: remove content validator about certification\n\nremove content validator about certification\n\nCloses-Bug: #2034990\nChange-Id: I42502b86b5361f129f4304b94ff294d3ea15c43f\n(cherry picked from commit 8166bade648597fa21121eb332c404f41b77a2d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7a516dfd8f940ea03610167b12a75f413d402a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7a516dfd8f940ea03610167b12a75f413d402a60"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
