)]}'
{"id":"openstack%2Fskyline-console~896738","triplet_id":"openstack%2Fskyline-console~master~Idb02ff0cce8ef48318b6eb9f0b5c8c6521fc961b","project":"openstack/skyline-console","branch":"master","topic":"data","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-09-29 13:38:58.000000000","reason":"Change was submitted"},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-09-29 11:58:18.000000000","reason":"\u003cGERRIT_ACCOUNT_28706\u003e replied on the change","reason_account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}},"hashtags":[],"change_id":"Idb02ff0cce8ef48318b6eb9f0b5c8c6521fc961b","subject":"feat: update data mapper in the base store","status":"MERGED","created":"2023-09-28 06:35:11.000000000","updated":"2023-09-29 13:40:01.000000000","submitted":"2023-09-29 13:38:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896738-data","meta_rev_id":"5dc748bbe5546754affce57c0f94dc5a2c08cc2f","_number":896738,"virtual_id_number":896738,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-29 13:38:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-09-29 11:58:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2023-09-28 08:07:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-09-29 11:58:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 07:44:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-28 08:07:14.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-09-29 11:38:31.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"7f8b8a1a5fd2e95f95754c8f929cb4e556d83d69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-28 06:35:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5c769d5b5454e0fdf6112dba9c79422af033251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 07:44:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b928d1a5ae1a4e0a9e15bc6f3a7e563a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/192324875ac34940b35e9ec36a00a0db : SUCCESS in 3m 54s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4e8fc459570f42df987809222f0d0f58 : SUCCESS in 11m 06s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1d2df2c9d0e748a1afa42b59ef404c62 : SUCCESS in 6m 57s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f512198411ab4f7182a91ece052c6285 : SUCCESS in 7m 32s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d04d448c461f4b0e89c795944e9957ce : SUCCESS in 1h 00m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a168ddf3584f4df799b5fb87120816b5 : SUCCESS in 59m 10s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/34aaeaea828c49d8a231756d6002f5f0 : SUCCESS in 59m 30s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d97c0c3ada204949b3ab939a2530174d : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8131e7a32595e58a95ec1cbe6b17bfe9e674e438","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-09-28 08:07:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ac808b6ecf85e8aabdba330e2d37bcf5d97c2d57","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-09-29 11:58:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31920ef86660cff9edbc9af89d3c803e8894756a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:58:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4e76962419597b132f883ac184ca53e8a974759","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 13:38:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42707090e6a74185a62db45970a7b19b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec05e58012bf41c0b116c836fda4244c : SUCCESS in 4m 27s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/cda76005f0ee4ee495ddc81105b5f099 : SUCCESS in 11m 38s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/3b1451565d774a66bc6d04b3aa8aa5ba : SUCCESS in 7m 34s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/c25d23bc8a0a4cde9104f266ceb1011f : SUCCESS in 7m 43s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/795e6fbb61ef4a2a8cd715582ff02a86 : SUCCESS in 1h 34m 48s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d373131ca493471491d6d7869abe239f : SUCCESS in 1h 12m 39s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/569d41c816a84301a434fffc05c55a10 : SUCCESS in 50m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f2f1b91c57954517931139b21dccdd61 : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":1},{"id":"26b22b499d25c7b17755e95b1209512fd4c31b09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 13:38:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5dc748bbe5546754affce57c0f94dc5a2c08cc2f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 13:40:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a378924a8545ed8f5de2e3f2ee9b8f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c7188de7c79446ca4fb0053cec270e2 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12bc328c2488801446591524e4f7b644bd6564ab","revisions":{"12bc328c2488801446591524e4f7b644bd6564ab":{"kind":"REWORK","_number":1,"created":"2023-09-28 06:35:11.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/38/896738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/38/896738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/896738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/896738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/896738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/38/896738/1"}}},"commit":{"parents":[{"commit":"8166bade648597fa21121eb332c404f41b77a2d8","subject":"fix: remove content validator about certification","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8166bade648597fa21121eb332c404f41b77a2d8"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-28 06:34:19.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-28 06:34:19.000000000","tz":480},"subject":"feat: update data mapper in the base store","message":"feat: update data mapper in the base store\n\nupdate the data mapper in the base store, add extra params for the function, which can be better rebuild the data used in the list page or detail page\n\nChange-Id: Idb02ff0cce8ef48318b6eb9f0b5c8c6521fc961b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12bc328c2488801446591524e4f7b644bd6564ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12bc328c2488801446591524e4f7b644bd6564ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
