)]}'
{"id":"openstack%2Fskyline-console~896752","triplet_id":"openstack%2Fskyline-console~master~If494648f4d42c6da54661400283968c7220bf2a4","project":"openstack/skyline-console","branch":"master","topic":"bug/2003057","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-10-07 10:42:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If494648f4d42c6da54661400283968c7220bf2a4","subject":"feat: support create non-bfv instance","status":"MERGED","created":"2023-09-28 09:21:14.000000000","updated":"2023-10-07 10:43:32.000000000","submitted":"2023-10-07 10:42:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896752-bug/2003057","meta_rev_id":"b0b8e822014664abf1f590fc646a1b1535b497d9","_number":896752,"virtual_id_number":896752,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-07 10:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-10-07 09:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-10-07 09:37:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 10:46:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-07 09:37:50.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"ef155d338a9a3ef4854e8953fd52e16bb996c135","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-28 09:21:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cc0bef9bf99a953b29ecd2e3764ead7af23deb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 10:46:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0eb900d07ff4446592d9a880796082e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1e0e00f1cc34d1fa94052a083691f34 : SUCCESS in 5m 12s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/80fb18719a93442991c582fd42476c37 : SUCCESS in 11m 07s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/3355b076383a402ba253233fc0c0e022 : SUCCESS in 5m 02s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/c576fa763f4445f1903eb3f4f3a3e10a : SUCCESS in 7m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/68d9483bbc374a77ab23bc2ab57666f1 : SUCCESS in 1h 22m 14s","accounts_in_message":[],"_revision_number":1},{"id":"84b6e28562eec2f2ad8a88b2b1312f7ee3d3bf16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-07 03:01:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"cfc979362f3920e15db3b6fbfe8f4dbf659eb2f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 03:46:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a83779b72b14eedbd2f876080d09faa\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fde7f769779c4c69b9e0561dff0bfba1 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aff43fc9d23f4b1296f8067348006715 : SUCCESS in 5m 20s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/75b0dc9f22c34b82930b5e7b2a75f638 : SUCCESS in 11m 38s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/3ecf4a7f1e16433992c55da19361b629 : SUCCESS in 6m 02s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2bf490ce1b614583a4d0cd918b828ca4 : SUCCESS in 6m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/141fdeb5d1884f84a1e11218d41cd7c1 : FAILURE in 36m 40s","accounts_in_message":[],"_revision_number":2},{"id":"a0933c43bb536359345d51d54e09db3360b92efb","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-07 07:54:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"98fa82f64ce368e74c4c56df2ddee7e66d7da268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 09:28:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4a95da44e2b437491ab2fe796e6713a\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f4b39fba2b94f2db3c76d8846bd6029 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e1068446d4449b9a92abd03e00f754 : SUCCESS in 5m 50s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c9807cdf23d34a9ba887302f49f55280 : SUCCESS in 11m 50s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4a33173946514884ad4c3abd122d6bc7 : SUCCESS in 4m 24s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/ffd0c9a26f4044d5a1c8316baa1bf75d : SUCCESS in 7m 19s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1d39f87d7e604eb2b38048332046c5db : SUCCESS in 1h 28m 34s","accounts_in_message":[],"_revision_number":2},{"id":"723a393145f7fa3913561e4a8a0ecca323d33354","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-10-07 09:37:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e91baf5b4d8b79d3b34217fc1ddd2c69a1930bf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 09:38:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2a4b16216d5d1e1d9593cdbe53b9e963747ea6fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 10:42:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be6e6187f1aa454fb2bccabc043c17a4\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eba3743f5bf54fb681f18cc58581c894 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5553c88c56584cb3971edea06efd330b : SUCCESS in 4m 20s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/dd0d016201394231b03cb65283214e5e : SUCCESS in 10m 50s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/fffba15bb24c4facb35cb5a57b634fb1 : SUCCESS in 7m 04s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/70ed5d96fa0845188506725ae511849d : SUCCESS in 5m 45s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4656820e3b554e39be29a70a3e93b84e : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":2},{"id":"5beeea55497525a4f51f0fa29b0045a304511697","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 10:42:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b0b8e822014664abf1f590fc646a1b1535b497d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 10:43:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df45ae942f804c7293a70ff138136bc5\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8834995ee7074e528b1f51f75090aaa9 : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a57f4ec851a24558b5a291bde134afd1 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a186ee521b0663cf528646140e82dcb34370fd4","revisions":{"fcaf140f20936e5a28db7a01cf8bdaf76b0c696f":{"kind":"REWORK","_number":1,"created":"2023-09-28 09:21:14.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/52/896752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/52/896752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/896752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/896752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/896752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/52/896752/1"}}},"commit":{"parents":[{"commit":"8166bade648597fa21121eb332c404f41b77a2d8","subject":"fix: remove content validator about certification","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8166bade648597fa21121eb332c404f41b77a2d8"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-28 09:18:39.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-28 09:20:14.000000000","tz":480},"subject":"feat: support create non-bfv instance","message":"feat: support create non-bfv instance\n\nsupport create non-bfv instance\n\nCloses-Bug: #2003057\nChange-Id: If494648f4d42c6da54661400283968c7220bf2a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fcaf140f20936e5a28db7a01cf8bdaf76b0c696f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fcaf140f20936e5a28db7a01cf8bdaf76b0c696f"}]},"branch":"refs/heads/master"},"6a186ee521b0663cf528646140e82dcb34370fd4":{"kind":"REWORK","_number":2,"created":"2023-10-07 03:01:40.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/52/896752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/52/896752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/896752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/896752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/52/896752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/52/896752/2"}}},"commit":{"parents":[{"commit":"12bc328c2488801446591524e4f7b644bd6564ab","subject":"feat: update data mapper in the base store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12bc328c2488801446591524e4f7b644bd6564ab"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-09-28 09:18:39.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-10-07 03:00:43.000000000","tz":480},"subject":"feat: support create non-bfv instance","message":"feat: support create non-bfv instance\n\nsupport create non-bfv instance\n\nCloses-Bug: #2003057\nChange-Id: If494648f4d42c6da54661400283968c7220bf2a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6a186ee521b0663cf528646140e82dcb34370fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6a186ee521b0663cf528646140e82dcb34370fd4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
