)]}'
{"id":"openstack%2Fskyline-console~896933","triplet_id":"openstack%2Fskyline-console~stable%2F2023.2~Ib824a9aa3a908c503dbe7379fcc99f0176b99582","project":"openstack/skyline-console","branch":"stable/2023.2","topic":"package","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-09-29 11:48:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib824a9aa3a908c503dbe7379fcc99f0176b99582","subject":"feat: update package for release","status":"MERGED","created":"2023-09-29 11:22:20.000000000","updated":"2023-09-29 11:49:11.000000000","submitted":"2023-09-29 11:48:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896933-package","meta_rev_id":"e82c036d2d3167e3b80018f26d857658882c8bbc","_number":896933,"virtual_id_number":896933,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-29 11:48:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-09-29 11:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-09-29 11:32:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-29 11:32:38.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2023-09-29 11:36:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"700fb4ce5a1414c7f1b48c5712ffff47774f87ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-09-29 11:22:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa228536551bfae3df395a3018f95c46177811dd","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-09-29 11:32:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"94c434befd245d72360105f16331caba4dbead24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:36:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cf6c050bb3d42d1bdaf7ff2be171d6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42ffb77dbbcd4e93b79e1e70f7a5fa19 : SUCCESS in 4m 31s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/2747832f311d4a7d974c3283d596e9d0 : SUCCESS in 10m 37s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/fee2b6692acd47a38b35bbb39ca8283a : SUCCESS in 7m 23s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/c1cb3fd0cf014c898cf2a13b15b11aeb : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0bd5f58d0320b34d7df1e461f9c61f89eab6c16d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:36:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b40f828ab854d696c80fb4c3f05f1025fbf632cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:48:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd8b36253efe4201b463a7b606abbbeb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/721020131234424097d907363d993e31 : SUCCESS in 4m 51s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9124f402b4dd47f5b470fbd558391013 : SUCCESS in 8m 42s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/03c1deaa9bfc4544bb38c75fac4b4076 : SUCCESS in 6m 29s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0acde010cc35459e9b7ff86e544a9796 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c6e50075cb7b212f15c31bdb1934efd5bb538539","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:48:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e82c036d2d3167e3b80018f26d857658882c8bbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:49:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9697cf038cfa4ea0b70ddfb6760341ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b6fcaf64bf4a3dbc242ec7fef757d1 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1044d4f8a8cbd8a579d793d3167c237bf080513","revisions":{"a1044d4f8a8cbd8a579d793d3167c237bf080513":{"kind":"REWORK","_number":1,"created":"2023-09-29 11:22:20.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/33/896933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/33/896933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/33/896933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/33/896933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/33/896933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/33/896933/1"}}},"commit":{"parents":[{"commit":"7a516dfd8f940ea03610167b12a75f413d402a60","subject":"fix: remove content validator about certification","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7a516dfd8f940ea03610167b12a75f413d402a60"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-09-29 11:20:09.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-09-29 11:20:09.000000000","tz":480},"subject":"feat: update package for release","message":"feat: update package for release\n\nupdate package for branch release\n\nChange-Id: Ib824a9aa3a908c503dbe7379fcc99f0176b99582\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a1044d4f8a8cbd8a579d793d3167c237bf080513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a1044d4f8a8cbd8a579d793d3167c237bf080513"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
