)]}'
{"id":"openstack%2Fskyline-console~896999","triplet_id":"openstack%2Fskyline-console~master~I3e680fc8810ac339bdf4ad6820b1dfe8d8d69d40","project":"openstack/skyline-console","branch":"master","attention_set":{"36242":{"account":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"last_update":"2023-10-02 09:01:25.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I3e680fc8810ac339bdf4ad6820b1dfe8d8d69d40","subject":"Added TOTP authentication","status":"NEW","created":"2023-10-02 07:52:12.000000000","updated":"2023-10-02 09:01:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"84b8cfb5a1c01236a96ac04c5850b4e1768d4357","_number":896999,"virtual_id_number":896999,"owner":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-02 09:01:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-02 09:01:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec48427b29d02c80b61a6d5396a4af3646221346","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"date":"2023-10-02 07:52:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84b8cfb5a1c01236a96ac04c5850b4e1768d4357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 09:01:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/adce2e83bcba460790e92e87e71ca502\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fb090d8aa8840c9ae0efaf2ba9e9a10 : SUCCESS in 3m 59s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/b13a0cc9ad5f47f69a004d1bd84aed28 : FAILURE in 7m 14s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/9661d491ebcc4653902df84fd1d9d519 : SUCCESS in 4m 53s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/44531121dbff40cfa6d50bfb021d298e : SUCCESS in 7m 00s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/77b1c72cf4404e2f8e92aff103150f1b : FAILURE in 1h 08m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"558723ce9bed9882b019929d59c4ed4491c790ef","revisions":{"558723ce9bed9882b019929d59c4ed4491c790ef":{"kind":"REWORK","_number":1,"created":"2023-10-02 07:52:12.000000000","uploader":{"_account_id":36242,"name":"omar asim mirza","email":"omarasimmirza@gmail.com","username":"omar.asim"},"ref":"refs/changes/99/896999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/99/896999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/896999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/896999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/896999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/99/896999/1"}}},"commit":{"parents":[{"commit":"e42b15d083d3df487a1461b3ea74e613a55a5630","subject":"Added TOTP authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e42b15d083d3df487a1461b3ea74e613a55a5630"}]}],"author":{"name":"omar asim mirza","email":"omarasimmirza@gmail.com","date":"2023-10-02 07:51:49.000000000","tz":0},"committer":{"name":"omar asim mirza","email":"omarasimmirza@gmail.com","date":"2023-10-02 07:51:49.000000000","tz":0},"subject":"Added TOTP authentication","message":"Added TOTP authentication\n\nChange-Id: I3e680fc8810ac339bdf4ad6820b1dfe8d8d69d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/558723ce9bed9882b019929d59c4ed4491c790ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/558723ce9bed9882b019929d59c4ed4491c790ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
