)]}'
{"id":"openstack%2Fskyline-console~897655","triplet_id":"openstack%2Fskyline-console~master~I8b061e1744c044f24e2704b7a241fc9ef1744c77","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-10-11 04:13:26.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-10-10 09:06:24.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-10-11 04:13:26.000000000","reason":"Change was submitted"},"34671":{"account":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"last_update":"2023-10-11 04:13:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b061e1744c044f24e2704b7a241fc9ef1744c77","subject":"Feat: Docker images to glance","status":"MERGED","created":"2023-10-09 07:47:39.000000000","updated":"2023-10-11 04:14:40.000000000","submitted":"2023-10-11 04:13:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897655","meta_rev_id":"d7ba8cc3104cc3a4b2e23a149a6072b0e4d08a97","_number":897655,"virtual_id_number":897655,"owner":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 04:13:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-10-11 02:40:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-10-11 02:40:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-09 09:26:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-09 12:57:30.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-10-09 12:57:30.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2023-10-09 12:57:30.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"794ca9c1ea739ea3064def58cea2ee9f4737cc77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-09 07:47:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a845057d9ec1c53d97964cf20c65cecc598f7326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-09 07:49:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0e210d169f95487d0bf24ced358269205b902b68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 09:26:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdb70d8bcf4d4ce2b4a8ad026a188896\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e869ae2373144d4b8fa3ba76455b3d3 : SUCCESS in 4m 30s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/5e983db1816c4d86821703c610179606 : SUCCESS in 7m 52s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/136b4e8b618a43f6b57e9a498660e4fc : SUCCESS in 8m 04s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/513a84dcbbd142d780daa85bc6eb9a8d : SUCCESS in 7m 28s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/dab7edd76a4a4a08bec3f769a67e7b5b : SUCCESS in 1h 31m 32s","accounts_in_message":[],"_revision_number":2},{"id":"64af958e6d58c404d1be7b7614776ca2e2e9b8ae","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-09 12:57:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"da2c17adc92f977faab953bb066583efbcf95088","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-10 08:15:13.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"15eeb64cf71df69a4efb2e1736fc84eb12cbfce9","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-10 08:17:56.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"0fa6820a63ef17dec0dbfa3d4ae082881c757a3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-10 08:43:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f39baf37d64469031b2aff5e3ba408a459d5b677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-10 08:46:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"848396c98a6f61a9fc0bca338c4c1281c573d5c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-10 08:48:14.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"36967912cf8931f6aa62687623c48aff0b38e2f5","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-10 08:48:59.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9feab55f5d81a16245297a333251e71907f6a0dc","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-10 09:06:24.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1656028644a3ea9923265243467a18a6df6c4dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 10:17:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c0117de8f4c4ecd932962b3ba52d1d7\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0662f2d550f4528ab1395edb5c3ade6 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c196cc57d11f43e1ae888b21b409c266 : SUCCESS in 5m 19s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ea4e32fa13b744b48f86fb9659fc335a : SUCCESS in 10m 35s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/37995690f0714ed4bef53c5758281170 : SUCCESS in 7m 30s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a223d2aef7704bdeb738f7a54993451f : SUCCESS in 7m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5966679561ac4c8095d4278f99a9fa0d : SUCCESS in 1h 24m 41s","accounts_in_message":[],"_revision_number":5},{"id":"4451d9c2d4b01eae83ae5e4d796c475d29b47350","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-11 02:40:06.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5c3cea0b6a0a1efffa529b91fd30b1d2793f60a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 02:40:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3c24bf74d4af681e5b5e3b97ff9b701c276bc462","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 04:13:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2a4300218bf4703940c827c49e8dfab\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45631609c3644cdd938269535d7fdd27 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5e686f2e5614efdaa8519f1305eb7e0 : SUCCESS in 4m 25s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0b62602a16c34e1bbfab5241b192f7ee : SUCCESS in 8m 30s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/9a94800b651c4cde83211669b97d1c8a : SUCCESS in 7m 31s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1a46b1cb6ce94184884a9c942415b400 : SUCCESS in 8m 48s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/79ab9f62ad1a4b6899f36dec124dde40 : SUCCESS in 1h 26m 12s","accounts_in_message":[],"_revision_number":5},{"id":"9982a0a44a5ce751df517938569a7da467daf4a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 04:13:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d7ba8cc3104cc3a4b2e23a149a6072b0e4d08a97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 04:14:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8733efd154dd484db0313b9cad51c25f\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8c119049a2749b798b0d3b6d25b1901 : SUCCESS in 46s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3867e3f0f6dc4fdc9af0c7606701a641 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a8e76700829cb1fb4ff45672e28d058d3e072de9","revisions":{"e6387f0ce3e4d986fecde2f734ba5f5696e8dd6e":{"kind":"REWORK","_number":1,"created":"2023-10-09 07:47:39.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/55/897655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/897655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/1"}}},"commit":{"parents":[{"commit":"43e93feec4499c1d12e6071e48adc99d4773a721","subject":"feat: update i18n for no-bfv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/43e93feec4499c1d12e6071e48adc99d4773a721"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-09 07:42:25.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-09 07:44:02.000000000","tz":0},"subject":"Feat: Docker images to glance","message":"Feat: Docker images to glance\n\nChange-Id: I8b061e1744c044f24e2704b7a241fc9ef1744c77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e6387f0ce3e4d986fecde2f734ba5f5696e8dd6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e6387f0ce3e4d986fecde2f734ba5f5696e8dd6e"}]},"branch":"refs/heads/master"},"19955f0f4ee9fcbe7d861afc98d72bee034723b8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-09 07:49:23.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/55/897655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/897655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/2"}}},"commit":{"parents":[{"commit":"43e93feec4499c1d12e6071e48adc99d4773a721","subject":"feat: update i18n for no-bfv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/43e93feec4499c1d12e6071e48adc99d4773a721"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-09 07:42:25.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-09 07:49:23.000000000","tz":0},"subject":"feat: Docker images to glance","message":"feat: Docker images to glance\n\nChange-Id: I8b061e1744c044f24e2704b7a241fc9ef1744c77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/19955f0f4ee9fcbe7d861afc98d72bee034723b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/19955f0f4ee9fcbe7d861afc98d72bee034723b8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c0fc1cbf8caa726094b55110a924d41c6638a36a":{"kind":"REWORK","_number":3,"created":"2023-10-10 08:43:29.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/55/897655/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/897655/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/3"}}},"commit":{"parents":[{"commit":"43e93feec4499c1d12e6071e48adc99d4773a721","subject":"feat: update i18n for no-bfv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/43e93feec4499c1d12e6071e48adc99d4773a721"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-09 07:42:25.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-10 08:39:37.000000000","tz":0},"subject":"Feat: Docker images to glance","message":"Feat: Docker images to glance\n\nChange-Id: I8b061e1744c044f24e2704b7a241fc9ef1744c77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c0fc1cbf8caa726094b55110a924d41c6638a36a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c0fc1cbf8caa726094b55110a924d41c6638a36a"}]},"branch":"refs/heads/master"},"099aea2187d16309757503a917822117da484aac":{"kind":"REWORK","_number":4,"created":"2023-10-10 08:46:19.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/55/897655/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/897655/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/4"}}},"commit":{"parents":[{"commit":"43e93feec4499c1d12e6071e48adc99d4773a721","subject":"feat: update i18n for no-bfv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/43e93feec4499c1d12e6071e48adc99d4773a721"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-09 07:42:25.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-10 08:42:32.000000000","tz":0},"subject":"Feat: Docker images to glance","message":"Feat: Docker images to glance\n\nChange-Id: I8b061e1744c044f24e2704b7a241fc9ef1744c77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/099aea2187d16309757503a917822117da484aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/099aea2187d16309757503a917822117da484aac"}]},"branch":"refs/heads/master"},"a8e76700829cb1fb4ff45672e28d058d3e072de9":{"kind":"REWORK","_number":5,"created":"2023-10-10 08:48:14.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/55/897655/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/897655/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/897655/5"}}},"commit":{"parents":[{"commit":"43e93feec4499c1d12e6071e48adc99d4773a721","subject":"feat: update i18n for no-bfv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/43e93feec4499c1d12e6071e48adc99d4773a721"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-09 07:42:25.000000000","tz":0},"committer":{"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-10 08:48:14.000000000","tz":0},"subject":"Feat: Docker images to glance","message":"Feat: Docker images to glance\n\nChange-Id: I8b061e1744c044f24e2704b7a241fc9ef1744c77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a8e76700829cb1fb4ff45672e28d058d3e072de9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a8e76700829cb1fb4ff45672e28d058d3e072de9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
