)]}'
{"id":"openstack%2Fskyline-console~897918","triplet_id":"openstack%2Fskyline-console~master~Ibbe4cf0ae22fa6523f1968ab99fab972c2e9331a","project":"openstack/skyline-console","branch":"master","topic":"monitor","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-10-11 07:56:35.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"Ibbe4cf0ae22fa6523f1968ab99fab972c2e9331a","subject":"fix: support custom fetch for detail monitor modal","status":"MERGED","created":"2023-10-11 04:41:35.000000000","updated":"2023-10-11 11:07:54.000000000","submitted":"2023-10-11 11:06:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897918-monitor","meta_rev_id":"bda903be3cb1277e201584ad66f519783894a558","_number":897918,"virtual_id_number":897918,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 11:06:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-10-11 06:16:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-10-11 06:16:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-11 06:01:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-11 06:16:03.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"8fb02dd3fc896e82eabaaac76b764209577289ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-11 04:41:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6e1d868abc0b72655c9e3c65b86a6a7a5a297cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 06:01:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3b87c85d89942f7820083fa08d84762\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698ca2dd4aee41959a5bdc62a7ec20ce : SUCCESS in 4m 44s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c6d8138081784e54a6f2b381f16d897a : SUCCESS in 11m 03s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e57aa60110d84b5b8d0ee1e58d567c65 : SUCCESS in 5m 49s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1dc376d86b924fae82ae2df33ba91c68 : SUCCESS in 7m 43s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/083b69d1d5ba4466a241f8e221274e31 : SUCCESS in 1h 17m 33s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/83096e0ecda041b68c5b317129ebae4c : SUCCESS in 1h 09m 30s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f412d7aa5830407b81e124c4a8290134 : SUCCESS in 46m 50s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4471785ffd1d4e8d9bed7ed3d5967eea : SUCCESS in 51m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c3db83893b41071b85f9666da98b8db67226bbc6","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-10-11 06:16:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bdb6e13a76941f17de5a0004c61065d4df36bb45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 06:16:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42f4e5dc925b8f5ad569baa8671d65f5ae42d3c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 07:54:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62fed6785a09466db56aa2600880117e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f730c0ec1c44e9ad1376ca4938a307 : SUCCESS in 4m 34s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/624eebd65b63456fa33fbaaf0a024874 : SUCCESS in 10m 17s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1b7aa285884744459ab0e1c727e98796 : SUCCESS in 4m 54s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f7261725e1f04618b2b2c2657a9c552f : SUCCESS in 5m 40s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/72497dabf4d843f7903730acde2feef1 : SUCCESS in 1h 31m 08s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9f384d69ea3140978652ca5635b3ac8b : SUCCESS in 55m 45s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/174a1e48052844e7ad355f2c621dd4be : SUCCESS in 1h 01m 42s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/85fe61f209574b6e810ed9c745f9bd0a : FAILURE in 1h 22m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4733cddac15f4a5d2bb0ada97da6c4cf3580ac90","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-11 07:56:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b299a9ed5f1efb8d8007e05367ebd51bcefad583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 09:32:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9366285a8284378994b5666723be051\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84ac73db19742d591eb298230eb25e3 : SUCCESS in 5m 06s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/390729df656149ee84900e5c1cc1188f : SUCCESS in 9m 24s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/17f9f050419b4a6ea3e01742c915a778 : SUCCESS in 7m 17s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/309afef352124b0f8736afa5709d63a8 : SUCCESS in 9m 13s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b9c1ae7853b74a6fa4bd8341ac2e0552 : SUCCESS in 1h 27m 32s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3c919e53cb3d40d281245c82c985ec7d : SUCCESS in 1h 13m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bc09ad4db7f54c49a3372826a9210bf0 : SUCCESS in 1h 01m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/191e406806934839b1721558608b7b2e : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":1},{"id":"bb225184147188471dbe7f530fbf99f0a56dcfbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 09:32:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58395db44fe8fd1467f186bc883897e8a3ec7b98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:06:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d20e8942b564bbd801d5c60dd1fe8c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4bf8a8d925f4bffbcb8c3dec70cbe9f : SUCCESS in 4m 59s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9e6605e2dd0949f0af7bde7cc256d8b3 : SUCCESS in 6m 52s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/59da9a06dd214ddfb45571226aab7dbd : SUCCESS in 7m 54s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a76e1ef76e75419abc198e9680a041ad : SUCCESS in 7m 45s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0fc0762bc7584bf98b8e5e5ec5d21775 : SUCCESS in 1h 33m 51s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bc41d494defd41edb19585b94a49855f : SUCCESS in 1h 03m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/dd1504c43a034c41b9f85486d7be3357 : SUCCESS in 49m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/47dba3883cb049ee92466568ace11cda : SUCCESS in 51m 03s","accounts_in_message":[],"_revision_number":1},{"id":"43beb2cfbeab30f5fb655be88f5e0f29bad2af62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:06:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bda903be3cb1277e201584ad66f519783894a558","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:07:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b53ff7d5f924c1c973b3b389e8ed2b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d46bc90a6a4defbca3984940bca7e9 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa829ea7c42264e08a728a0554c0b08be22cfe6c","revisions":{"fa829ea7c42264e08a728a0554c0b08be22cfe6c":{"kind":"REWORK","_number":1,"created":"2023-10-11 04:41:35.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/18/897918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/18/897918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/897918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/897918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/897918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/18/897918/1"}}},"commit":{"parents":[{"commit":"a8e76700829cb1fb4ff45672e28d058d3e072de9","subject":"Feat: Docker images to glance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a8e76700829cb1fb4ff45672e28d058d3e072de9"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-10-11 04:38:10.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-10-11 04:39:58.000000000","tz":480},"subject":"fix: support custom fetch for detail monitor modal","message":"fix: support custom fetch for detail monitor modal\n\nsupport use custom fetch when click the extend icon in the monitor chart card\n\nChange-Id: Ibbe4cf0ae22fa6523f1968ab99fab972c2e9331a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fa829ea7c42264e08a728a0554c0b08be22cfe6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fa829ea7c42264e08a728a0554c0b08be22cfe6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
