)]}'
{"id":"openstack%2Fskyline-console~898055","triplet_id":"openstack%2Fskyline-console~master~I8c5387b4e03edfb5f94764a7bb7c60759962871e","project":"openstack/skyline-console","branch":"master","topic":"image","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-10-12 06:54:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8c5387b4e03edfb5f94764a7bb7c60759962871e","subject":"refactor: update columns function for image","status":"MERGED","created":"2023-10-12 03:34:16.000000000","updated":"2023-10-12 06:55:00.000000000","submitted":"2023-10-12 06:54:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898055-image","meta_rev_id":"2fec8061b4b7921d8d7873d6feafb747aa27dd53","_number":898055,"virtual_id_number":898055,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-12 06:54:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-10-12 05:50:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-10-12 05:50:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-12 04:44:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 05:50:51.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"28cab782203b6392aaf7dd5e5250641a89585b6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-12 03:34:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b857b83700563e45569f0549e66cdbd254e4401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 04:44:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c95b3f3c06324764b6ec0ac83b69fb8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa7512aea714e5cbb8c024eb5fb212a : SUCCESS in 4m 40s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/5666f1988c8b44e5825eb0a7faae0887 : SUCCESS in 10m 55s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e05af5089b41432080b3fd743eeee7f5 : SUCCESS in 8m 00s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1203641c42794436a63a2a6f10996ba8 : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a47aadad848840d39d0d27c54ae032d4 : SUCCESS in 1h 08m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a8468534f7d946c2780ad7c6519c41008450ad86","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-10-12 05:50:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f43b288d71590f7e6e5d1caf09b071d1eced7991","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 05:51:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"834934c810e178196f9825c90d1c303313481881","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 06:54:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f45a845d059d4a14b1fd56ddcb290152\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06ec996986d74d078ab05095012d66ae : SUCCESS in 5m 16s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/00ce13206c674efaa0baa7929688085b : SUCCESS in 11m 21s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/9bb1ff15d00149c386ca8f4edfce32ed : SUCCESS in 7m 03s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/9215fe41b89b482aaaa7bf02eaf8965f : SUCCESS in 5m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f7fdaab350e646bdbbcebbfcb1db9633 : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6ee7bf2ee79cf1823014ae4d4b431aebb008693d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 06:54:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2fec8061b4b7921d8d7873d6feafb747aa27dd53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 06:55:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfc46e37a8354bdd9271d3d827d6c79c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3914d1d2f9bb4fda9680ada348f2f7b7 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d56f8ea6ce98e536ff9488b252aae1fb8b046289","revisions":{"d56f8ea6ce98e536ff9488b252aae1fb8b046289":{"kind":"REWORK","_number":1,"created":"2023-10-12 03:34:16.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/55/898055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/898055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/898055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/898055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/898055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/898055/1"}}},"commit":{"parents":[{"commit":"c2a80fda7acc6290bbd53f10676f64ca6acca120","subject":"Merge \"feat: add container format column for image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c2a80fda7acc6290bbd53f10676f64ca6acca120"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-10-12 03:31:26.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-10-12 03:31:34.000000000","tz":480},"subject":"refactor: update columns function for image","message":"refactor: update columns function for image\n\nupdate getColumns function for image, then other class can better update the columns by inheritance class\n\nChange-Id: I8c5387b4e03edfb5f94764a7bb7c60759962871e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d56f8ea6ce98e536ff9488b252aae1fb8b046289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d56f8ea6ce98e536ff9488b252aae1fb8b046289"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
