)]}'
{"id":"openstack%2Fskyline-console~898085","triplet_id":"openstack%2Fskyline-console~master~I2ba707e78dfb1dbac6706186bce312b9e8873868","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-10-16 06:47:56.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"34671":{"account":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"last_update":"2023-10-16 09:46:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2ba707e78dfb1dbac6706186bce312b9e8873868","subject":"feat: Add Keypair column list to instances.","status":"MERGED","created":"2023-10-12 06:17:11.000000000","updated":"2023-10-16 09:47:54.000000000","submitted":"2023-10-16 09:46:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898085","meta_rev_id":"7bf2e415342ce595a46802dd3ba42b1ba56b8145","_number":898085,"virtual_id_number":898085,"owner":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 09:46:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-10-16 08:12:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-10-16 08:12:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}],"CC":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-12 06:47:19.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"CC"},{"updated":"2023-10-12 07:34:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 11:34:28.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"CC"},{"updated":"2023-10-12 11:34:28.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"CC"},{"updated":"2023-10-16 01:31:20.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"45590d9e9c1ca10372c354a1282959c510eddcd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-12 06:17:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1506cbdccd2e43b005382a3973084f79fc785921","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-12 06:47:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bfcf07e52f77c08756043daa218fa56840bd32f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 07:34:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/397874f8f8344951bf071bb0152f0426\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5071e353c17341bcba271ec3621def54 : SUCCESS in 3m 56s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c27611d0065b4249a4818eb76598e052 : SUCCESS in 6m 40s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0a171d09cae041af8997b453849244d7 : SUCCESS in 7m 38s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/deedd2b8f9b4403bb907c502feb3a7eb : SUCCESS in 7m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/81f7048c0c40417eb5c4f39c1d51cf9c : FAILURE in 1h 09m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6b6bf1f82f3276c57f6500ad6e32196134f24d86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-12 08:40:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"861ac4c35072c7b3f44062f437807846fb52a53a","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-12 08:41:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3efb5126d0a89ffeb7c0636beff906a844de18f8","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-12 11:34:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"47ab61fb82ad69b91b536926cfcdd5e9ddafdaed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 11:49:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bcb53ea31ca4032960eace7b24c33aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb43a184e5bc4106aac051d35ba83774 : SUCCESS in 4m 39s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/34005b55765c4cf2937c8e470854c7fe : SUCCESS in 10m 36s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/37d1989ea5114524b825be93fe1558da : SUCCESS in 7m 29s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/170af1d80a3f44fb818c7ea5b0dca861 : SUCCESS in 7m 35s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/29adf1f01b364f3998b75f8a62bb24a2 : TIMED_OUT in 3h 04m 40s","accounts_in_message":[],"_revision_number":2},{"id":"fecf4678c3db85f504dbf30392ffceefe1e19073","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-12 11:55:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a21b38dedcdf6952ea3aefdd5f283d5921fe5327","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 13:35:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c85d15f690784e61954f9611d45552d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6500e4775c894383bd4630da47930e2d : SUCCESS in 4m 31s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/dda5990787454e69b06de794d0cabdf7 : SUCCESS in 7m 11s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e01d67f1b3a14a7c95a078703a6943ee : SUCCESS in 6m 24s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/124df53a12b64cc185f3df132c5d61c8 : SUCCESS in 5m 02s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/06ff1f9ecf474d4a85b002ab099def62 : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":2},{"id":"fc47fa3f02d0b8d5cfd82d715de5c3fe141cf699","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-16 01:31:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"64eab9a424007df456cc756a315afa5dc0a69375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-16 06:18:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f7d25b6e1e3c5dff22038b73e75087ea4db31c10","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-16 06:20:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0311c813e61a5a883ea62ed57fd6462cadc334ab","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-16 06:47:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"06e49671220739929557a8b4954466f207017f8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 07:55:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ab55358a99240f398c2118aca7765f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09337503e354c2eb331326c41e060b5 : SUCCESS in 4m 09s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/5a0dc6a89ce44ef3bba71c289e6745b9 : SUCCESS in 6m 58s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/d66ff078d0c44b929c1636c4419cb561 : SUCCESS in 7m 20s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/30538872965b43b0af43a1646b7c1e1b : SUCCESS in 7m 40s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/af00b0999d504ee2bdbc84a2455493b0 : SUCCESS in 1h 31m 00s","accounts_in_message":[],"_revision_number":3},{"id":"cf5312a9c0cf7d10344a14dcdc5926c36f21c2e8","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-16 08:12:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"69eea97f07aaa449f09e7a9bc86c1b814c4d7201","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 08:13:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5b57984cfa2d49e69256498b449800022e0434d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:46:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77488e1bde104d5e8976ad31426d4dc1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4397c64c243b432790fdffca21aad3c1 : SUCCESS in 4m 19s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/52d557eead3d4f82ab46817b72121a75 : SUCCESS in 8m 34s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/63c88e68182b4f668949f260fa9e81ea : SUCCESS in 6m 02s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b576dc31d0364adab92338ef8c9990d5 : SUCCESS in 7m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/835786a5ed6f4f179e6231aaeb65f06a : SUCCESS in 1h 33m 22s","accounts_in_message":[],"_revision_number":3},{"id":"4b0de2a549b765da864585b7bddf6ba4d74e6a4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:46:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7bf2e415342ce595a46802dd3ba42b1ba56b8145","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:47:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/209695e8f4e84fc8b043686128cb71f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b4b86d951c047cfbbb1cde5ef01069a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b06fc9a3cc4c1bae08ea7fff88de3680f24bf873","revisions":{"43ae045006c8a2479147d9228517b67b46908eba":{"kind":"REWORK","_number":1,"created":"2023-10-12 06:17:11.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/85/898085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/898085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/898085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/898085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/898085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/898085/1"}}},"commit":{"parents":[{"commit":"c2a80fda7acc6290bbd53f10676f64ca6acca120","subject":"Merge \"feat: add container format column for image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c2a80fda7acc6290bbd53f10676f64ca6acca120"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-12 06:12:00.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-12 06:13:32.000000000","tz":0},"subject":"feat: Add Keypair column list to instances.","message":"feat: Add Keypair column list to instances.\n\nChange-Id: I2ba707e78dfb1dbac6706186bce312b9e8873868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/43ae045006c8a2479147d9228517b67b46908eba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/43ae045006c8a2479147d9228517b67b46908eba"}]},"branch":"refs/heads/master"},"d394693f0a8f41224920525d1c585d26af49c2ee":{"kind":"REWORK","_number":2,"created":"2023-10-12 08:40:58.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/85/898085/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/898085/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/898085/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/898085/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/898085/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/898085/2"}}},"commit":{"parents":[{"commit":"c2a80fda7acc6290bbd53f10676f64ca6acca120","subject":"Merge \"feat: add container format column for image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c2a80fda7acc6290bbd53f10676f64ca6acca120"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-12 06:12:00.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-12 08:37:01.000000000","tz":0},"subject":"feat: Add Keypair column list to instances.","message":"feat: Add Keypair column list to instances.\n\nChange-Id: I2ba707e78dfb1dbac6706186bce312b9e8873868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d394693f0a8f41224920525d1c585d26af49c2ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d394693f0a8f41224920525d1c585d26af49c2ee"}]},"branch":"refs/heads/master"},"b06fc9a3cc4c1bae08ea7fff88de3680f24bf873":{"kind":"REWORK","_number":3,"created":"2023-10-16 06:18:01.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/85/898085/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/898085/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/898085/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/898085/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/898085/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/898085/3"}}},"commit":{"parents":[{"commit":"c2a80fda7acc6290bbd53f10676f64ca6acca120","subject":"Merge \"feat: add container format column for image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c2a80fda7acc6290bbd53f10676f64ca6acca120"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-12 06:12:00.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-16 06:14:07.000000000","tz":0},"subject":"feat: Add Keypair column list to instances.","message":"feat: Add Keypair column list to instances.\n\nChange-Id: I2ba707e78dfb1dbac6706186bce312b9e8873868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b06fc9a3cc4c1bae08ea7fff88de3680f24bf873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b06fc9a3cc4c1bae08ea7fff88de3680f24bf873"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
