)]}'
{"id":"openstack%2Fskyline-console~898308","triplet_id":"openstack%2Fskyline-console~master~Ieb5023f8596c1c401ad2d0868c99f0155c1c45a8","project":"openstack/skyline-console","branch":"master","topic":"designate","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-10-26 09:20:40.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-10-26 09:20:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieb5023f8596c1c401ad2d0868c99f0155c1c45a8","subject":"feat: update dns zone","status":"MERGED","created":"2023-10-16 08:16:06.000000000","updated":"2023-10-26 09:21:45.000000000","submitted":"2023-10-26 09:20:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898308-designate","meta_rev_id":"6084bfcb55259a654c9e4d0cda018f11dc36e5ef","_number":898308,"virtual_id_number":898308,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-26 09:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-10-26 07:44:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-10-26 07:44:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-16 09:40:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-25 11:34:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-10-26 07:44:12.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"603ee72c51bccae26c86792a6663eeb34b668e42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-16 08:16:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3161f5141de9304e28aefac8d4f3ee9e3bca061","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:40:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3f3095b0b3f465d959ddaa90e735d77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d98e3103223845379ff0524540419a52 : SUCCESS in 4m 32s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/2887631282d346918cfe4a2f056d9025 : SUCCESS in 10m 25s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0b80cc7494204f5f8dae188cfdcc0cf9 : SUCCESS in 4m 59s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b417da522b9843aca172fbfbcbc8dafb : SUCCESS in 8m 12s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/68277703ce9f4a2785e21ee2e575ba72 : SUCCESS in 1h 07m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5be6da91667740f6af976615629c1b16 : SUCCESS in 1h 09m 08s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/efc2904c129249539c622e2f11453cb4 : SUCCESS in 55m 58s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5697d185b6744d7fae996812e2593d19 : FAILURE in 1h 16m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5b56b476361b639e2abe9147089cb02a065fbbec","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-16 09:56:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d0177c4b9212c738b69c2f63294ccd8b95f41bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 11:27:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc8e7b0b35d34fc5908d79344a7bfd63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee19b427aae4580b01410a72af797c0 : SUCCESS in 6m 03s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/6a9137872d3f450f837547b1a2e4a94e : SUCCESS in 11m 32s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/5061b1c39ae14bcbb21f7b3434569efa : SUCCESS in 7m 30s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/8fdfc03374b74bb988523fb5af526289 : SUCCESS in 6m 05s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/dfc502293f8b4bf4b7afdf0bd49c4d33 : SUCCESS in 1h 30m 12s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e202fbc2296746c3b6c100f32381479a : SUCCESS in 1h 08m 08s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2fa95169ce51475599f223bfedb5250d : SUCCESS in 44m 21s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/99a92d726b8c48cba04fbac8236e71db : FAILURE in 1h 24m 54s","accounts_in_message":[],"_revision_number":1},{"id":"bc7bc2278ec7df4efb7cafe68efd1f07386949d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-25 04:41:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"772ab06ef8b97c0c3974334e0cdf17b693e0d539","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 06:09:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ae7fa6691b142dba5f93883c9dd0e88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/199b9375f7744d3bb1e98d208e33bdf9 : SUCCESS in 4m 15s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/407658d420fd417995617758452d6baf : SUCCESS in 10m 55s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1d5289d9e78e4fbfa7e7f9f95ad34c35 : SUCCESS in 7m 33s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/845a332175264e5dad33c6f8e79bf980 : SUCCESS in 5m 29s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/741e4c9e6f4a40a8966870a9e40d76a1 : SUCCESS in 1h 00m 37s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d6813bb17152488fbdb055ce7652e109 : SUCCESS in 46m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/705017e8b5d64598879efdbea64b99c2 : SUCCESS in 39m 00s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0ae5a9912fc74a6ebcdd31d5d750b1cb : FAILURE in 1h 21m 03s","accounts_in_message":[],"_revision_number":2},{"id":"8b33325aab057e4513ae12965ef5585901e1ea16","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-25 11:34:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ef67757adab11a1791fee4452d1db7c002d14077","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-26 03:33:39.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3def27cb2018f9fcee324e6ade630f2821eae95f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-26 04:13:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"58989a6454192e09d6ad736f015a0a984f2a9c9f","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-26 04:14:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7d55f8a7444c09da42c5848f5b7a92ff17766923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 05:40:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/714e5785febe402393027b753ba7bd8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b719e6f175e458a837bb9c5e5eddc7c : SUCCESS in 4m 40s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/57f5675ca9a945aea77419e19016eedb : SUCCESS in 8m 54s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ab1b875f4c24449ca5c7f1422a50c721 : SUCCESS in 7m 36s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/53f4dd58851b4e8fb1a253ca3627b23e : SUCCESS in 4m 48s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d90bd1d3536f4c7bb7560998c969d501 : FAILURE in 1h 24m 24s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0ff857788aee46dfb001d0d3367eb617 : SUCCESS in 1h 07m 10s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b335416252af4e59ae4ecf038646acee : SUCCESS in 39m 16s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/45b3e52575654661880af3b643430ab1 : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":4},{"id":"2e3dd00c5df63fae15a3c205c687eddf034797f6","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-26 05:45:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e07ca4d5f463a4d4010d4e93efd295e04691c9e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 07:36:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d273dfa355b64402a253870f3ccd6fbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c006b76c95e04faca50c99800acf23ff : SUCCESS in 5m 35s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0494ef717fa541d0b4d0d3f4e9ad9431 : SUCCESS in 11m 52s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/446e1f912c1c4f8ba27145f7586ca603 : SUCCESS in 8m 36s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b5b756fb061d401497877fbe762f09ca : SUCCESS in 7m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/28f353a144b34664bd77735437051667 : SUCCESS in 1h 41m 35s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c7a61dbfc8fb40b9ae029bbba1b522fa : SUCCESS in 1h 01m 42s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/570e2d59c4614c97aa557e15d5639544 : SUCCESS in 43m 14s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a033e16f83764db894b2cf8ecd601782 : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":4},{"id":"860dd9eba9dbd8a8fd59b10e2bcb1fe90132262f","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-10-26 07:44:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"eb8a2519cd68fd50372916949a86e0a67e1a7dab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 07:44:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b5ac4201e3f7b1061cace64334490fad072fb68f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 09:20:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c03d7e12103f4efbb3cf48fa87a4bf5c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925ea9895891477b9de84acea95bad9e : SUCCESS in 4m 51s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4608012707c34e59899069ce0e1fa2e5 : SUCCESS in 8m 46s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ac45dea786184b4d8b6b6a04284d0a83 : SUCCESS in 7m 45s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/fb09ff0418804cd798f0e0589d9cc55f : SUCCESS in 5m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5b6856a9346747989342439a4e21bde1 : SUCCESS in 1h 30m 42s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/850507fde3eb499aaaad79703ac673a2 : SUCCESS in 1h 04m 47s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/17e052939d4c413e9c3d2fefcf5b7a88 : SUCCESS in 50m 11s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9caa389ca4df4a8e8bb1c054abd8621b : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":4},{"id":"949eade17cb5ef745a6b483bbc9082914619e3df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 09:20:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6084bfcb55259a654c9e4d0cda018f11dc36e5ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 09:21:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6563d7258aff4c01b3a44563f41f36f0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/323b832ade204e7aa79d32dec9a88647 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"28c14c426cfb81a31ba00b12de41a173a61aa9bd","revisions":{"b5fb1f3b91939a3716b3cfae4498211803e7822a":{"kind":"REWORK","_number":1,"created":"2023-10-16 08:16:06.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/08/898308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/08/898308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/1"}}},"commit":{"parents":[{"commit":"d56f8ea6ce98e536ff9488b252aae1fb8b046289","subject":"refactor: update columns function for image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d56f8ea6ce98e536ff9488b252aae1fb8b046289"}]}],"author":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-10-16 08:15:36.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2023-10-16 08:15:36.000000000","tz":480},"subject":"feat: update dns zone","message":"feat: update dns zone\n\n1. fix dns zone api fetch in the zone list page\n2. udpate create dns zone\n3. update dns zone list\n4. update dns zone detail\n\nCloses-Bug: #2032857\nChange-Id: Ieb5023f8596c1c401ad2d0868c99f0155c1c45a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b5fb1f3b91939a3716b3cfae4498211803e7822a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b5fb1f3b91939a3716b3cfae4498211803e7822a"}]},"branch":"refs/heads/master"},"49349bcd078866d32fd9c6f906a0bb3487855c2d":{"kind":"REWORK","_number":2,"created":"2023-10-25 04:41:03.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/08/898308/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/08/898308/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/2"}}},"commit":{"parents":[{"commit":"065242761ad9b1909e5bc29d16127bdaa221aecc","subject":"Do not add all_projects \u003d true when fetching ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/065242761ad9b1909e5bc29d16127bdaa221aecc"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-25 04:40:33.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-25 04:40:33.000000000","tz":480},"subject":"feat: update dns zone","message":"feat: update dns zone\n\n1. fix dns zone api fetch in the zone list page\n2. udpate create dns zone\n3. update dns zone list\n4. update dns zone detail\n\nCloses-Bug: #2032857\nChange-Id: Ieb5023f8596c1c401ad2d0868c99f0155c1c45a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/49349bcd078866d32fd9c6f906a0bb3487855c2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/49349bcd078866d32fd9c6f906a0bb3487855c2d"}]},"branch":"refs/heads/master"},"1842dd3f683610e871560a607898ff9c4738e76c":{"kind":"REWORK","_number":3,"created":"2023-10-26 03:33:39.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/08/898308/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/08/898308/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/3"}}},"commit":{"parents":[{"commit":"521b965a239024b7b24698ef6492c37b2e8491d3","subject":"fix: remove volume create image case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/521b965a239024b7b24698ef6492c37b2e8491d3"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-25 04:40:33.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-26 03:33:21.000000000","tz":480},"subject":"feat: update dns zone","message":"feat: update dns zone\n\n1. fix dns zone api fetch in the zone list page\n2. udpate create dns zone\n3. update dns zone list\n4. update dns zone detail\n5. update edit dns zone\n6. update delete dns zone\n\nCloses-Bug: #2032857\nChange-Id: Ieb5023f8596c1c401ad2d0868c99f0155c1c45a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1842dd3f683610e871560a607898ff9c4738e76c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1842dd3f683610e871560a607898ff9c4738e76c"}]},"branch":"refs/heads/master"},"28c14c426cfb81a31ba00b12de41a173a61aa9bd":{"kind":"REWORK","_number":4,"created":"2023-10-26 04:13:50.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/08/898308/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/08/898308/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/08/898308/4"}}},"commit":{"parents":[{"commit":"521b965a239024b7b24698ef6492c37b2e8491d3","subject":"fix: remove volume create image case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/521b965a239024b7b24698ef6492c37b2e8491d3"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-25 04:40:33.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-26 04:13:31.000000000","tz":480},"subject":"feat: update dns zone","message":"feat: update dns zone\n\n1. fix dns zone api fetch in the zone list page\n2. udpate create dns zone\n3. update dns zone list\n4. update dns zone detail\n5. update edit dns zone\n6. update delete dns zone\n\nCloses-Bug: #2032857\nChange-Id: Ieb5023f8596c1c401ad2d0868c99f0155c1c45a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/28c14c426cfb81a31ba00b12de41a173a61aa9bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/28c14c426cfb81a31ba00b12de41a173a61aa9bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
