)]}'
{"id":"openstack%2Fskyline-console~898316","triplet_id":"openstack%2Fskyline-console~master~I6acd1fa5f2c8bb817f29f0e11b0908cf73c7653c","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-10-18 08:56:43.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-10-18 07:44:49.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-10-18 08:56:43.000000000","reason":"Change was submitted"},"34671":{"account":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"last_update":"2023-10-18 08:56:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6acd1fa5f2c8bb817f29f0e11b0908cf73c7653c","subject":"Do not add all_projects \u003d true when fetching ports","status":"MERGED","created":"2023-10-16 12:03:05.000000000","updated":"2023-10-18 08:57:46.000000000","submitted":"2023-10-18 08:56:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898316","meta_rev_id":"2df7634d3f72b3526a9b01d7aa0446175ebb0e5b","_number":898316,"virtual_id_number":898316,"owner":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-18 08:56:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-10-18 07:44:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-10-18 07:44:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-16 13:03:14.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2023-10-16 13:03:14.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-10-16 13:03:14.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2023-10-16 13:42:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d7757f906c6d7a72bb673ff18bc42977e49573a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-16 12:03:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c23c353ec85d95dbcf1c6dfe86fe9b25f92b7b55","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-16 13:03:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c38e16aeb686667bc06e25b568c7b8af4e5a1c1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 13:42:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91fd22a9ceb74b539291a8eaf6307d46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da34b41f43b0420a92815e403c0627d0 : SUCCESS in 4m 37s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f15f96a45f0548e8b7cc3eca8d39c942 : SUCCESS in 6m 53s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/f8101e0706c94eb99910f82310b0ed65 : SUCCESS in 7m 30s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b8340380d7104e5b99326f0d932fa8a9 : SUCCESS in 4m 50s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ac9c6c1b896f43a5b43cd38936da29bd : SUCCESS in 1h 32m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c9a77b9be81a88f4e54d58986160d49d81aded31","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-17 01:07:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1596879df81cab0cbeb0b205892ed9ca3fd33242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-17 06:21:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2aa4082c995d33c3b723c0b574aec54bed64f4d5","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2023-10-17 06:21:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21525876f28963c9dd1fe208e35a31368f4b9570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 07:16:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d402f3754ed4b9fa7bede311ed0e12a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f12f5d72404770ab7396c273e9cac4 : SUCCESS in 4m 44s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/fc029fa25f094ff49f11ef27d405b1d9 : SUCCESS in 7m 37s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/90cbf62c46d94cea969a0ef504c58bec : SUCCESS in 5m 48s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/6fd585c5a4ed4e3394cc98e92b247c4e : SUCCESS in 6m 21s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f81b6004f6464ff4b26903d1dc75859e : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":2},{"id":"221c9973a9b84c06587bb207521bbefad4ebd1d9","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-18 07:44:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b590f192a9db7ddaf7075b9036aad5a7a7b70d34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 07:45:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6765301afdb3ac87c29f4b024916d36847e9ba4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 08:56:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4a427c908344e968f95d11e776eb7dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b46b8b20bc94452a9665c7f0d806c608 : SUCCESS in 4m 50s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/38cc1098df144d89af8cba8f2681e34c : SUCCESS in 11m 53s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/579414be987a4220b2d1eddd542c8cd2 : SUCCESS in 7m 39s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a5d912f24e3744d8a22ab5ed15f5541e : SUCCESS in 8m 31s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/23c012c7538d46698f10a865e330f7da : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":2},{"id":"d34c0bbef13c3b6366bc31aed993a9af57b99ec6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 08:56:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2df7634d3f72b3526a9b01d7aa0446175ebb0e5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 08:57:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/791b69ae750d4dca87bbe856c5520240\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39c7b33ded4c4fed801462cda81aa62d : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"065242761ad9b1909e5bc29d16127bdaa221aecc","revisions":{"d53885ca3c8436650b01c4e1b74f0cc6cf38feae":{"kind":"REWORK","_number":1,"created":"2023-10-16 12:03:05.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/16/898316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/16/898316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/898316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/898316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/898316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/16/898316/1"}}},"commit":{"parents":[{"commit":"8be87ab24e5c21934f06ae2277f5b5004a593f91","subject":"Merge \"feat: Add Keypair column list to instances.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8be87ab24e5c21934f06ae2277f5b5004a593f91"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-16 11:50:38.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-16 11:59:25.000000000","tz":0},"subject":"Do not add all_projects \u003d true when fetching ports","message":"Do not add all_projects \u003d true when fetching ports\n\nWhen go to the instance detail page it tries to fetch ports\n\nwith all_projects \u003d true but its API does not contain all_projects\n\nChange-Id: I6acd1fa5f2c8bb817f29f0e11b0908cf73c7653c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d53885ca3c8436650b01c4e1b74f0cc6cf38feae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d53885ca3c8436650b01c4e1b74f0cc6cf38feae"}]},"branch":"refs/heads/master"},"065242761ad9b1909e5bc29d16127bdaa221aecc":{"kind":"REWORK","_number":2,"created":"2023-10-17 06:21:06.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/16/898316/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/16/898316/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/898316/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/898316/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/898316/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/16/898316/2"}}},"commit":{"parents":[{"commit":"8be87ab24e5c21934f06ae2277f5b5004a593f91","subject":"Merge \"feat: Add Keypair column list to instances.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8be87ab24e5c21934f06ae2277f5b5004a593f91"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-16 11:50:38.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2023-10-17 06:17:11.000000000","tz":0},"subject":"Do not add all_projects \u003d true when fetching ports","message":"Do not add all_projects \u003d true when fetching ports\n\nWhen go to the instance detail page it tries to fetch ports\n\nwith all_projects \u003d true but its API does not contain all_projects\n\nChange-Id: I6acd1fa5f2c8bb817f29f0e11b0908cf73c7653c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/065242761ad9b1909e5bc29d16127bdaa221aecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/065242761ad9b1909e5bc29d16127bdaa221aecc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
