)]}'
{"id":"openstack%2Fskyline-console~899438","triplet_id":"openstack%2Fskyline-console~master~I918c1abdba059866ac86f53b5e7e2b7fbdfdfc98","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-10-27 02:53:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I918c1abdba059866ac86f53b5e7e2b7fbdfdfc98","subject":"fix: Values in the hypervisor were not matched","status":"MERGED","created":"2023-10-27 00:09:47.000000000","updated":"2023-10-27 02:55:39.000000000","submitted":"2023-10-27 02:53:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899438","meta_rev_id":"eeef5fa8ba12215d371e84d15bb90b04c40a19e1","_number":899438,"virtual_id_number":899438,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-27 02:53:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-10-27 01:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-10-27 01:43:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-27 01:17:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-27 01:43:05.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"74794424099a79ac4bfc9d0c9abac88e35e9c280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-10-27 00:09:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19b830cfef8d66f4eece6d7ba001de9abde32f65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 01:17:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b8dceb3179541feb90af9bbdbe8d448\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21b56bb44127483d8e133b6c1d5918ac : SUCCESS in 4m 31s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f86376419e3e4157a313dc45e9750fbb : SUCCESS in 7m 37s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4b0bd7f993ce4b6aac54fd4bc8805708 : SUCCESS in 5m 18s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/deacfb37d8234aaa8846e0cad62b654c : SUCCESS in 5m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2a578bebd3104936962dd1146a7f2f9b : SUCCESS in 1h 01m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a84f5d998aed009f12d5fdf96f98eda302790869","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-10-27 01:43:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a3e7ba0b8b4f67ac52ab22151c1b9ac8feb6fdef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 01:43:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c905379724f525cc7cc4d1fd6cf3aa9646daf630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 02:53:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a296076cbbc54b159c26845468c5f26e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23ceeaaa6c854c2894bef96c1dd25d90 : SUCCESS in 4m 38s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d1c5a913d4e548a0a7bedf34d115ac43 : SUCCESS in 10m 44s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/67a212eda7374d66912a738d26c93ae0 : SUCCESS in 4m 53s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/98723a8a059f4c40ab940e2489759247 : SUCCESS in 4m 35s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/48f64e1f5fef49cf92eefc19f73bfa16 : SUCCESS in 1h 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6611e01b87f758a070b742fdc819e4fb2c74e745","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 02:53:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eeef5fa8ba12215d371e84d15bb90b04c40a19e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 02:55:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b2318efe5b147a2967e29976e1b0769\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fe481892b4a4f80903fa29f21918a5e : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af9f6b74cc18e9e785c959811ead0cd3d0599df9","revisions":{"af9f6b74cc18e9e785c959811ead0cd3d0599df9":{"kind":"REWORK","_number":1,"created":"2023-10-27 00:09:47.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/38/899438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/38/899438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/899438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/899438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/899438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/38/899438/1"}}},"commit":{"parents":[{"commit":"91aa92b55af25968ccaaf5c655a1ce84fee8c1f2","subject":"feat: update record set of dns zone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/91aa92b55af25968ccaaf5c655a1ce84fee8c1f2"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-10-26 07:01:25.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-10-26 09:52:18.000000000","tz":480},"subject":"fix: Values in the hypervisor were not matched","message":"fix: Values in the hypervisor were not matched\n\nIronic node has no vcpu and memory usages data\n\nCloses-Bug: #2040971\nChange-Id: I918c1abdba059866ac86f53b5e7e2b7fbdfdfc98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/af9f6b74cc18e9e785c959811ead0cd3d0599df9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/af9f6b74cc18e9e785c959811ead0cd3d0599df9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
