)]}'
{"id":"openstack%2Fskyline-console~899566","triplet_id":"openstack%2Fskyline-console~master~Ib0b1d36fbaee52d3054da826f56e6b5f85a28457","project":"openstack/skyline-console","branch":"master","topic":"instance","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-10-31 16:02:28.000000000","reason":"Change was submitted"},"34680":{"account":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"last_update":"2023-10-31 16:02:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib0b1d36fbaee52d3054da826f56e6b5f85a28457","subject":"fix: fix system disk size check","status":"MERGED","created":"2023-10-30 07:00:46.000000000","updated":"2023-10-31 16:04:07.000000000","submitted":"2023-10-31 16:02:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899566-instance","meta_rev_id":"3fa73cf1c9842469fc51649da753733c8c27b378","_number":899566,"virtual_id_number":899566,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-31 16:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-10-31 14:35:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-10-31 14:35:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}],"CC":[{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-30 08:16:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-30 08:21:19.000000000","updated_by":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"reviewer":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"state":"CC"},{"updated":"2023-10-31 14:35:19.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"6e8db9ea4980b0103555b4fde8d0bcd67a8ecc9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-30 07:00:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"589970fdbda5a5b3f428ee4edba26fdbdf126733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 08:16:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/83de9d5ec7624daa9ec7bf728ee00e95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da5dfc2cf59649389a316630e016114d : SUCCESS in 5m 22s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8be6135d846147399bcdf4f82ecb2b04 : SUCCESS in 9m 00s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/9a0bcf7cd8714203a49e039cd88cb2be : SUCCESS in 7m 14s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/c6333c07efb341569a4e91522d131b37 : SUCCESS in 7m 01s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0f4f876c25954b7a9f90f8888d70846c : FAILURE in 1h 14m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5aa70355a3cbf57029fb970ec6bb1f35eb913e35","author":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"date":"2023-10-30 08:21:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8812909ef8e0c97a4d07e43ea2d64b0929954c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-30 09:15:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c568f3027fd2e1012b9fb85b33eaa404d7b6ae52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-30 09:24:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"73f15ea051cf9d6e4a64cd14aa1f5122645a7164","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-30 09:25:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b92959fc35f5ac7f2f52573cee0d85ed0e13827d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 10:40:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d3621dc720041d3b0bafe6106c0347b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e58176f8ad3463eb12a47349426e594 : SUCCESS in 4m 19s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c4cbc7b268a04204b56f0d62543cf3d0 : SUCCESS in 8m 25s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/8e044e8c77eb4290b450c5d11705c89b : SUCCESS in 5m 48s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/7c60f4eebce843268b51162a34f45957 : SUCCESS in 6m 01s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/af85be04d969401b9b344e1192a85964 : FAILURE in 1h 13m 50s","accounts_in_message":[],"_revision_number":3},{"id":"bfe7c6ea07642190d2f753e5f094b1c660a0bdc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-31 09:08:34.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"947f5379d5c9d07708ff2ce6c536f8107a1ab272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 10:24:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c597edf6363f4cb0ba9cfb7ff5901e35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16be120bd4c9445a82e82a7b238fc81c : SUCCESS in 4m 20s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c0982e7441c643178ab9c6fac6f6f5ff : SUCCESS in 8m 43s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7f56fe664a46482590603f998b236cb6 : SUCCESS in 6m 46s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/9e7e554910a3495b8c6bfe28c1021f16 : SUCCESS in 7m 05s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ffa353eb6a6248c5b472fb263c11aa9b : SUCCESS in 1h 10m 56s","accounts_in_message":[],"_revision_number":4},{"id":"7268860aa6dde980a77127566c133f90e4c6b3bf","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-10-31 14:35:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1370f642aa3cbd754471da62098bbbff4314ff67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 14:35:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"92b84983bd963ca2e2cae23cbf4e58776014fc9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 16:02:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db0b484a65d54ab0a80f10d5a8639da5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78b31dc9285248ae9d75a47e0e26deb6 : SUCCESS in 4m 55s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/b895046cd7394829befaef58549ef38d : SUCCESS in 8m 51s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/deed29fd64954b3d9837695f54ad8010 : SUCCESS in 6m 13s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f498ead05b024acfaf6e1924000b55e2 : SUCCESS in 6m 48s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/493d51446192411aaf141daeb67a7c4f : SUCCESS in 1h 16m 46s","accounts_in_message":[],"_revision_number":4},{"id":"46b330d8e3fab6ee4a94077a377153329bb98f00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 16:02:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3fa73cf1c9842469fc51649da753733c8c27b378","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 16:04:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd7e2765685b4bd287a7c62fad6c45a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/078c477dee39437e99079f7907eaa440 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2af37719c1eaf38b7f1140aabd01210fd1a94be9","revisions":{"e90f481f3203ef44e467e1ebb036c82747f99a6e":{"kind":"REWORK","_number":1,"created":"2023-10-30 07:00:46.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/66/899566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/66/899566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/1"}}},"commit":{"parents":[{"commit":"f7ded6afc7e9eacd5e2b737416b8cc08ff3d27d0","subject":"feat: support dns zone and recordsets in administrator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f7ded6afc7e9eacd5e2b737416b8cc08ff3d27d0"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-30 06:42:46.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-30 07:00:07.000000000","tz":480},"subject":"fix: fix system disk size check","message":"fix: fix system disk size check\n\nAdd system disk size check when create instance.\n\nCloses-Bug: #2038269\nChange-Id: Ib0b1d36fbaee52d3054da826f56e6b5f85a28457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e90f481f3203ef44e467e1ebb036c82747f99a6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e90f481f3203ef44e467e1ebb036c82747f99a6e"}]},"branch":"refs/heads/master"},"5635ca3e582938b7e95db5edf1f41eb19b4ae36e":{"kind":"REWORK","_number":2,"created":"2023-10-30 09:15:51.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/66/899566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/66/899566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/2"}}},"commit":{"parents":[{"commit":"f7ded6afc7e9eacd5e2b737416b8cc08ff3d27d0","subject":"feat: support dns zone and recordsets in administrator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f7ded6afc7e9eacd5e2b737416b8cc08ff3d27d0"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-30 06:42:46.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-30 09:15:39.000000000","tz":480},"subject":"fix: fix system disk size check","message":"fix: fix system disk size check\n\nAdd system disk size check when create instance.\n\nCloses-Bug: #2038269\nChange-Id: Ib0b1d36fbaee52d3054da826f56e6b5f85a28457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5635ca3e582938b7e95db5edf1f41eb19b4ae36e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5635ca3e582938b7e95db5edf1f41eb19b4ae36e"}]},"branch":"refs/heads/master"},"acc2adbd40243c528a0b3dee30ca52feb4a2eb03":{"kind":"REWORK","_number":3,"created":"2023-10-30 09:24:33.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/66/899566/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/66/899566/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/3"}}},"commit":{"parents":[{"commit":"f7ded6afc7e9eacd5e2b737416b8cc08ff3d27d0","subject":"feat: support dns zone and recordsets in administrator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f7ded6afc7e9eacd5e2b737416b8cc08ff3d27d0"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-30 06:42:46.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-30 09:24:17.000000000","tz":480},"subject":"fix: fix system disk size check","message":"fix: fix system disk size check\n\nAdd system disk size check when create instance.\n\nCloses-Bug: #2038269\nChange-Id: Ib0b1d36fbaee52d3054da826f56e6b5f85a28457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/acc2adbd40243c528a0b3dee30ca52feb4a2eb03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/acc2adbd40243c528a0b3dee30ca52feb4a2eb03"}]},"branch":"refs/heads/master"},"2af37719c1eaf38b7f1140aabd01210fd1a94be9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-10-31 09:08:34.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/66/899566/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/66/899566/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/66/899566/4"}}},"commit":{"parents":[{"commit":"e25560d3a6ea2510e6dd700067af1ea2dd1d4491","subject":"fix: fix instance stop e2e case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e25560d3a6ea2510e6dd700067af1ea2dd1d4491"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-30 06:42:46.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-31 09:08:22.000000000","tz":480},"subject":"fix: fix system disk size check","message":"fix: fix system disk size check\n\nAdd system disk size check when create instance.\n\nCloses-Bug: #2038269\nChange-Id: Ib0b1d36fbaee52d3054da826f56e6b5f85a28457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2af37719c1eaf38b7f1140aabd01210fd1a94be9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2af37719c1eaf38b7f1140aabd01210fd1a94be9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
