)]}'
{"id":"openstack%2Fskyline-console~899639","triplet_id":"openstack%2Fskyline-console~master~I04be0a63157d6fb7493062e872170db17ff5043f","project":"openstack/skyline-console","branch":"master","topic":"e2e","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-10-31 08:12:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I04be0a63157d6fb7493062e872170db17ff5043f","subject":"fix: fix instance stop e2e case","status":"MERGED","created":"2023-10-31 01:34:29.000000000","updated":"2023-10-31 08:14:02.000000000","submitted":"2023-10-31 08:12:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899639-e2e","meta_rev_id":"770a04077a1dd8bbf0d0604315f6378691087d08","_number":899639,"virtual_id_number":899639,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-31 08:12:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-10-31 07:08:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-10-31 07:08:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-31 02:44:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-31 07:08:07.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"a6a847576cd1a3b0ff2dad10ddab35c76e8b2408","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-31 01:34:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78a1bf875943d5a45d5fa95e94f8b2a7d1b51ff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 02:44:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5608932198994058b505aef9e1c0ffe5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9e5a9eb43184e028deb54d83106f00e : SUCCESS in 3m 58s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/1d77a180bf4e41688ab735306fa5b6e0 : SUCCESS in 7m 44s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/a18fe7a3a1b04920b11c5d8c1302e9d5 : SUCCESS in 4m 44s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5c205e51291e4fd69efd700c882bfd01 : SUCCESS in 4m 46s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d7c5b620abbe49dbaeecbb4c7e653f92 : FAILURE in 1h 09m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3870fb06e1f81d7ba43b6ac1c634ea38f3014c8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-10-31 03:26:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3a982bbf28a9e67f10ae041d2be8ad872636d032","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 06:44:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37dda80e7e6a477aa7e152b45d3c18aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4a18b2605d54f6880014e1d80fcc1d3 : SUCCESS in 5m 20s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/402e85c141a04d72bb42edc10173c95b : SUCCESS in 8m 41s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c9ad0dd1d0cd40ab8f06ba99d3570bc5 : SUCCESS in 7m 19s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1294cd051b2c40849e8073e05a637881 : SUCCESS in 6m 53s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e5b0f0adcc704c46a0d08dfb1a6b7fbb : SUCCESS in 1h 11m 24s","accounts_in_message":[],"_revision_number":2},{"id":"900b2e8db87ce262c4ec227f15f4adda18dbfe14","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-10-31 07:08:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a7c2f16ecc73f836dc2b73590e0c32cc15f276bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 07:08:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1e92153f51b5021daefca80902de6d3011058e2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 08:12:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a919373f8d314bf4b322ff57904c53a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/943309a358fa453ca624301d393b5b9c : SUCCESS in 4m 12s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/03f6a30c383c4d6fae32c76fe4c63929 : SUCCESS in 8m 26s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0f3a995b567d482382c9519048bfbd27 : SUCCESS in 7m 18s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a11162877860482a895a46b058ab5360 : SUCCESS in 4m 52s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6fac2ab9cb7740ea90c5925fba348950 : SUCCESS in 1h 02m 46s","accounts_in_message":[],"_revision_number":2},{"id":"f33552cdc045ba16120c42af3c891c613d73e562","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 08:12:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"770a04077a1dd8bbf0d0604315f6378691087d08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 08:14:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a3e7bda8f244b64a9b9ec78b78722c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ffde53fbf6e48c78360e5c7e1c42285 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e25560d3a6ea2510e6dd700067af1ea2dd1d4491","revisions":{"deefa10cfef8294f94a5d5c024e9bc558f230407":{"kind":"REWORK","_number":1,"created":"2023-10-31 01:34:29.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/39/899639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/39/899639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/39/899639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/39/899639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/39/899639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/39/899639/1"}}},"commit":{"parents":[{"commit":"f7ded6afc7e9eacd5e2b737416b8cc08ff3d27d0","subject":"feat: support dns zone and recordsets in administrator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f7ded6afc7e9eacd5e2b737416b8cc08ff3d27d0"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-31 01:33:22.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-31 01:34:13.000000000","tz":480},"subject":"fix: fix instance stop e2e case","message":"fix: fix instance stop e2e case\n\nFix the stop e2e case of the instance, which always gets a failed result.\n\nChange-Id: I04be0a63157d6fb7493062e872170db17ff5043f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/deefa10cfef8294f94a5d5c024e9bc558f230407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/deefa10cfef8294f94a5d5c024e9bc558f230407"}]},"branch":"refs/heads/master"},"e25560d3a6ea2510e6dd700067af1ea2dd1d4491":{"kind":"REWORK","_number":2,"created":"2023-10-31 03:26:09.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/39/899639/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/39/899639/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/39/899639/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/39/899639/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/39/899639/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/39/899639/2"}}},"commit":{"parents":[{"commit":"f7ded6afc7e9eacd5e2b737416b8cc08ff3d27d0","subject":"feat: support dns zone and recordsets in administrator","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f7ded6afc7e9eacd5e2b737416b8cc08ff3d27d0"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-31 01:33:22.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-10-31 03:25:47.000000000","tz":480},"subject":"fix: fix instance stop e2e case","message":"fix: fix instance stop e2e case\n\nFix the stop e2e case of the instance, which always gets a failed result.\n\nChange-Id: I04be0a63157d6fb7493062e872170db17ff5043f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e25560d3a6ea2510e6dd700067af1ea2dd1d4491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e25560d3a6ea2510e6dd700067af1ea2dd1d4491"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
