)]}'
{"id":"openstack%2Fskyline-console~900493","triplet_id":"openstack%2Fskyline-console~master~I357b4464aad23da3901b05c9940672a04b0a1861","project":"openstack/skyline-console","branch":"master","topic":"actions","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-11-09 03:58:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I357b4464aad23da3901b05c9940672a04b0a1861","subject":"fix: fix project name check","status":"MERGED","created":"2023-11-09 01:47:37.000000000","updated":"2023-11-09 03:59:23.000000000","submitted":"2023-11-09 03:58:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900493-actions","meta_rev_id":"3a81e57e923ca9ea2ef15eaf344e5a0c218fc4ce","_number":900493,"virtual_id_number":900493,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-09 03:58:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-11-09 02:48:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-11-09 02:48:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 02:37:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-09 02:48:10.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"456e0d441ad280ff89fd80be29ecc822f7b086b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-09 01:47:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c255046edeba2558f58d7a638311afc8d4ad0ff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 02:37:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21c66ac967644916beaa4c015faa65d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b729b2eca72437ea851eb050f0fb9cb : SUCCESS in 4m 44s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d249c5aba3cb4342b7125f11dad752b1 : SUCCESS in 11m 13s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1a42240f972c466e802be72689c27fe2 : SUCCESS in 7m 23s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/63e98851a24f431fb6f65435f4c64449 : SUCCESS in 9m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0306ef9feecd414998a5b89f5f95dbcd : SUCCESS in 49m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b37ee88c0037e0852b4841cba2f1ddb8e9ac9c24","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-09 02:48:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28dc5d26d7b54439b8b4380be5c13f740d3a137d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 02:48:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c7a6381c752063bdbea40d12fde19f998a8a8df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 03:58:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2cb4cd4841c4543acc199d89842c96a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c095d25c80e8474b91755f8f00124ff5 : SUCCESS in 5m 28s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ef1ffb48e75f4a35a4299a3ce548e5d2 : SUCCESS in 12m 25s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/67766f7c519f4890951716ea599a6bc8 : SUCCESS in 8m 41s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/7610f558fa494bb8942bfe68c04c392d : SUCCESS in 8m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7d61bc0032ba4401bf79c62cbdd9cc9f : SUCCESS in 1h 01m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d59b133970bdf95ee475975bd6ed8b52fb3fa656","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 03:58:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3a81e57e923ca9ea2ef15eaf344e5a0c218fc4ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 03:59:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef939a8594524a5e9a68d14c29cbeb5c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3be9aef41142409c856d297e5436572f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2075632e36b797d8e8fd95c59b2b01a2c537155d","revisions":{"2075632e36b797d8e8fd95c59b2b01a2c537155d":{"kind":"REWORK","_number":1,"created":"2023-11-09 01:47:37.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/93/900493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/900493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/900493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/900493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/900493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/900493/1"}}},"commit":{"parents":[{"commit":"2d6eeec7962bb6dd976c38629a926705258b2ebe","subject":"feat: update modal action for batch items","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2d6eeec7962bb6dd976c38629a926705258b2ebe"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-09 01:47:05.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-09 01:47:05.000000000","tz":480},"subject":"fix: fix project name check","message":"fix: fix project name check\n\nProject names in the same domain can not be repeated, fix the name check when creating and editing project\n\nChange-Id: I357b4464aad23da3901b05c9940672a04b0a1861\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2075632e36b797d8e8fd95c59b2b01a2c537155d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2075632e36b797d8e8fd95c59b2b01a2c537155d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
