)]}'
{"id":"openstack%2Fskyline-console~900494","triplet_id":"openstack%2Fskyline-console~master~I2cc2f6c800d85619cb9f3ca12e2e4b7d93c52125","project":"openstack/skyline-console","branch":"master","topic":"actions","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-11-09 04:35:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2cc2f6c800d85619cb9f3ca12e2e4b7d93c52125","subject":"fix: fix username check","status":"MERGED","created":"2023-11-09 01:55:42.000000000","updated":"2023-11-09 04:36:30.000000000","submitted":"2023-11-09 04:35:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900494-actions","meta_rev_id":"d23231889e845059f37c3d5cda299a2797e2a96e","_number":900494,"virtual_id_number":900494,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-09 04:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-11-09 03:29:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-11-09 03:29:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 02:47:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-09 03:29:57.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"a98ac8a63977047a4385f9b1a1fab4d94b60e985","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-09 01:55:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27526e91e3a8a7ede6650d86f7da9d5758d29f36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 02:47:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f7750d175c34d3f99c40c6405748051\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4be37fe6f54a449ea2ea29c202115a56 : SUCCESS in 6m 00s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/434b6462692b4a51b5c95e4d9656b834 : SUCCESS in 6m 38s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/2fea4e90b0554f31b813a115ec957df7 : SUCCESS in 8m 07s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/3ceab07650da43038aa7181074c1b026 : SUCCESS in 8m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/735b32125fed4a7fa84aad86690056dc : SUCCESS in 49m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1c0a76abc770277041df5cd382eb5a90885bdaa7","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-09 03:29:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e8f2b27be09cc8b4e1ab0707bc6612d6d886491b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 03:30:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2400553ed03e6e835ef6917bb5430b30cbb179f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 04:35:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e716e91a2cf047878c1d80e2115092f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a00d1353fd04b76b4e36fc1049bbbb2 : SUCCESS in 4m 29s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8a38497fa31e45e0a9d979d7eedc9462 : SUCCESS in 10m 25s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/23e476a20c21491a9a314ab98339dd98 : SUCCESS in 7m 23s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/390ad128262c40eeaafc910f84f1e3d1 : SUCCESS in 7m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/eee17d094a6e42c78b2ad22c0b387483 : SUCCESS in 59m 48s","accounts_in_message":[],"_revision_number":1},{"id":"900db8d7ae985e6b7e4fd853536e3d532f14a282","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 04:35:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d23231889e845059f37c3d5cda299a2797e2a96e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 04:36:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fa066afbf2f4b5f9151901749152836\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33374910d121482a976f127eea5435f4 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"663f8a26659e4328a068a11dc0f8300e5ea32d0c","revisions":{"663f8a26659e4328a068a11dc0f8300e5ea32d0c":{"kind":"REWORK","_number":1,"created":"2023-11-09 01:55:42.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/94/900494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/94/900494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/900494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/900494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/94/900494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/94/900494/1"}}},"commit":{"parents":[{"commit":"2075632e36b797d8e8fd95c59b2b01a2c537155d","subject":"fix: fix project name check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2075632e36b797d8e8fd95c59b2b01a2c537155d"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-09 01:55:17.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-09 01:55:17.000000000","tz":480},"subject":"fix: fix username check","message":"fix: fix username check\n\nUser names in the domain can not be repeated, check the name check when creating and editing user\n\nChange-Id: I2cc2f6c800d85619cb9f3ca12e2e4b7d93c52125\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/663f8a26659e4328a068a11dc0f8300e5ea32d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/663f8a26659e4328a068a11dc0f8300e5ea32d0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
