)]}'
{"id":"openstack%2Fskyline-console~900495","triplet_id":"openstack%2Fskyline-console~master~I174318b9b3c3dd6639a96a0f45e6d5a9935c06f8","project":"openstack/skyline-console","branch":"master","topic":"actions","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-11-09 06:15:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I174318b9b3c3dd6639a96a0f45e6d5a9935c06f8","subject":"fix: fix user group name check","status":"MERGED","created":"2023-11-09 02:14:51.000000000","updated":"2023-11-09 06:16:11.000000000","submitted":"2023-11-09 06:15:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900495-actions","meta_rev_id":"8bdc34b124459c4826f5bada3ccb1f4f89cbe363","_number":900495,"virtual_id_number":900495,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-09 06:15:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-11-09 05:28:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-11-09 05:28:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 03:57:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-09 05:28:32.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"a01ac2b0314f47b613a66a42e1ecb30f09caed62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-09 02:14:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36317085d8ddc96eed6ee7bbdfcb333b8543fa92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 03:57:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ab349cbd323459880777edbf9f0a835\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7640304d203f45db9e63c93ee91405f0 : SUCCESS in 5m 33s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/fddaf46244254f2382d6974f1676d9b8 : SUCCESS in 10m 56s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1a1d0fb1ce0744c6bd43d695c97bc852 : SUCCESS in 7m 53s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f605f1ec75d9430eba333adcb5766f2f : SUCCESS in 8m 11s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3843bf0638094b819abe24eb4a19ae6e : SUCCESS in 54m 34s","accounts_in_message":[],"_revision_number":1},{"id":"224c586cd4ab91c48a5ee2915385661313f47206","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-09 05:28:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0bebe3c74789505599796ddaecdc7ba50356026b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 05:29:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7af85fd438d111150b2e6647d33ae644cc3d8187","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 06:15:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a4aebe12532496d8d2021e947cbfe99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fcae994b63342d182ca13e548ec9c93 : SUCCESS in 5m 02s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/98e432628c9c484db14a7fea9e63de6d : SUCCESS in 10m 03s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/afc5d3306781418294f07ff836b91a25 : SUCCESS in 7m 30s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f128b09c5a884e2c8ffacd509b43a07a : SUCCESS in 7m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f3abe2ac52aa4a25bdb3356b3a92ead9 : SUCCESS in 44m 48s","accounts_in_message":[],"_revision_number":1},{"id":"31d195568f7b8d199533bf966542f2fa2add285b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 06:15:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8bdc34b124459c4826f5bada3ccb1f4f89cbe363","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 06:16:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2e04358ae214bb7b6ed0380e37b2caa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f063ec25ba0446785c11ed9662b7c0a : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b9c5248a9834a5dd87dc974ca407c59c62d6d41","revisions":{"9b9c5248a9834a5dd87dc974ca407c59c62d6d41":{"kind":"REWORK","_number":1,"created":"2023-11-09 02:14:51.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/95/900495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/95/900495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/95/900495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/95/900495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/95/900495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/95/900495/1"}}},"commit":{"parents":[{"commit":"663f8a26659e4328a068a11dc0f8300e5ea32d0c","subject":"fix: fix username check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/663f8a26659e4328a068a11dc0f8300e5ea32d0c"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-09 02:13:40.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-09 02:13:40.000000000","tz":480},"subject":"fix: fix user group name check","message":"fix: fix user group name check\n\nUser group names in the domain can not be repeated, update the name check when creating and editing user group\n\nChange-Id: I174318b9b3c3dd6639a96a0f45e6d5a9935c06f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9b9c5248a9834a5dd87dc974ca407c59c62d6d41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9b9c5248a9834a5dd87dc974ca407c59c62d6d41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
