)]}'
{"id":"openstack%2Fskyline-console~900851","triplet_id":"openstack%2Fskyline-console~master~I80a500c6d67b51203c1d2e5868f94c4800c72714","project":"openstack/skyline-console","branch":"master","topic":"lb","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-11-14 09:05:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I80a500c6d67b51203c1d2e5868f94c4800c72714","subject":"feat: support editing pool admin_state_up","status":"MERGED","created":"2023-11-14 06:21:47.000000000","updated":"2023-11-14 09:06:40.000000000","submitted":"2023-11-14 09:05:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900851-lb","meta_rev_id":"ca486da4850204523b8295ebd4c69a29dced8294","_number":900851,"virtual_id_number":900851,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-14 09:05:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-11-14 07:47:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-11-14 07:47:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-14 07:43:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-14 07:47:47.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"00096acfe9206c901ffa28b13a616835753d14ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-14 06:21:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56f3f8c56b3cf756cb2874fb790356494d935f2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 07:43:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6718c89735354521850743908bd75be8\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72162da9cd0e400e9e58988a16aafd83 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f5fa5dbdad94a50add387bb18fae272 : SUCCESS in 4m 15s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/465a59c52a8d47c8b8686b46231e87e2 : SUCCESS in 10m 37s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/347ce204bac84a9b847565d68f76beb0 : SUCCESS in 7m 05s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/49c90914abeb47958f2ddb1e207425f1 : SUCCESS in 7m 54s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4ee5fb4bd49b422799d17afb68b11b39 : SUCCESS in 1h 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c690d353ebf57b5c36d717e5ee5f486d7252ff2f","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-14 07:47:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e15adef8366ae966229c7199b486cc43ad3eee0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 07:48:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7f6f6c7f70c2a57ac3dc418d828598d2a946fe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 09:05:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b902494d069a433f99bb52f3d774ceea\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/171a3582e8604c44a1f552fd341b54b7 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aa48872ec844cbda6d46c61ea65661c : SUCCESS in 4m 13s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/6ee59fb28dee474caebd65b450fe441b : SUCCESS in 10m 40s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e060588925144153b327d4d6e4cd6222 : SUCCESS in 4m 36s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f2fe2135c79b4fd7ab142f8ccbacca2b : SUCCESS in 4m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d2a899854b53429da8ff6d4ba9f793d1 : SUCCESS in 1h 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6b96532b9279ef057b1a68fd96fc779d7cec5961","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 09:05:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ca486da4850204523b8295ebd4c69a29dced8294","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 09:06:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbc60ef1622d42aca0723ce1f5a79a6d\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdf1531263644e4998ffb729ff7bdc0c : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a06d3626e884aefade68c66212fce49 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"772b1b5d460a0c4067e353b874861ca75ea8b729","revisions":{"772b1b5d460a0c4067e353b874861ca75ea8b729":{"kind":"REWORK","_number":1,"created":"2023-11-14 06:21:47.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/51/900851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/900851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/900851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/900851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/900851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/900851/1"}}},"commit":{"parents":[{"commit":"081ff7ee8bd2fea02d7f1f830240a420a7c1a138","subject":"feat: support insert_headers for listener","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/081ff7ee8bd2fea02d7f1f830240a420a7c1a138"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-14 06:17:11.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-14 06:20:51.000000000","tz":480},"subject":"feat: support editing pool admin_state_up","message":"feat: support editing pool admin_state_up\n\n1. support editing admin_state_up when edit listener\u0027s pool of lb\n2. support display admin_state_up info in the pool info card in the listener detail page\n3. update lb algorithm display in the pool info card in the listener detail page\n\nChange-Id: I80a500c6d67b51203c1d2e5868f94c4800c72714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/772b1b5d460a0c4067e353b874861ca75ea8b729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/772b1b5d460a0c4067e353b874861ca75ea8b729"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
