)]}'
{"id":"openstack%2Fskyline-console~900974","triplet_id":"openstack%2Fskyline-console~master~Id1a9c32ddae09b1ae3dfb613bd7468a372aa5478","project":"openstack/skyline-console","branch":"master","topic":"lb","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-11-15 06:15:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1a9c32ddae09b1ae3dfb613bd7468a372aa5478","subject":"feat: update edit health monitor","status":"MERGED","created":"2023-11-15 02:54:17.000000000","updated":"2023-11-15 06:16:17.000000000","submitted":"2023-11-15 06:15:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900974-lb","meta_rev_id":"1cbf4936594b4ec01d697f65c4c0c5f528d3a314","_number":900974,"virtual_id_number":900974,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-15 06:15:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-11-15 05:31:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-11-15 05:31:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-15 04:48:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-15 05:31:22.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"a4e1c390a9194db0a619b435aa0d082aafd9e7f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-15 02:54:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b5dc7469e70b5b92cb46f7993fa9256fa161f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 04:48:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/379bad7d814e49e69da5ff9d662eb413\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8d903e9a8b441a58b834019cb9735fe : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b857d6c07d1046ddab545c687094a53a : SUCCESS in 4m 47s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c869312753514a2bab2d169632cbd539 : SUCCESS in 11m 40s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e793808d6b7f44f0a47ed69f302bcb3d : SUCCESS in 5m 59s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1a1c3ac66f1e436e87eb8f254afbcbe6 : SUCCESS in 7m 33s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1b87f600e8334e05a3b6d31308f944f9 : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":1},{"id":"de185268f1cc338942f5734ce7ef5c49e87f5811","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-15 05:31:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6ee8baf5ceb23c4f236a0d057f3d3405b331a978","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 05:31:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"257fd269d1688fe82169f68fe6a8b273129f3dfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 06:15:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/992886b608c9456ebc322fa441a5dfb6\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d044eaeef2c49c284bc654037195f2c : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aced4ee9431d4363a6826a009030f139 : SUCCESS in 4m 17s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/403b4caddb784302be9bac206c4c2a37 : SUCCESS in 10m 42s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/62ed4f22e86d46e39a618aa0cf2157e5 : SUCCESS in 4m 25s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/733266f9fcfc4b1fa93c2426551d0e39 : SUCCESS in 5m 47s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/226b8a67395b4e28850edbf23835246a : SUCCESS in 42m 45s","accounts_in_message":[],"_revision_number":1},{"id":"72589ffe6b0b11fb44fb34547dd073b63ad56523","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 06:15:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1cbf4936594b4ec01d697f65c4c0c5f528d3a314","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 06:16:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd3104e853e74ab8910a900463e4dffb\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b37e0634ec1a417c83d1b30b6b3b2ec4 : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7cffbbeab0f481ab8f6ed17181cb623 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2801a7913aa27f445f51126156494463f19e814","revisions":{"b2801a7913aa27f445f51126156494463f19e814":{"kind":"REWORK","_number":1,"created":"2023-11-15 02:54:17.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/74/900974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/74/900974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/900974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/900974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/900974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/74/900974/1"}}},"commit":{"parents":[{"commit":"73653da708eba0cbb38f680fd1ee0cf0433ab509","subject":"feat: update editing admin_state_up for listener","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/73653da708eba0cbb38f680fd1ee0cf0433ab509"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-15 02:47:27.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-15 02:53:48.000000000","tz":480},"subject":"feat: update edit health monitor","message":"feat: update edit health monitor\n\n1. refactor listener detail and actions code to optimize requests\n2. fix remove health monitor\n3. hide admin_state_up form item when disable health monitor in the lb step create page\n4. support updating admin_state_up when editing health monitor\n5. support admin_state_up info in the health monitor card in the lisenter detail page\n\nChange-Id: Id1a9c32ddae09b1ae3dfb613bd7468a372aa5478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b2801a7913aa27f445f51126156494463f19e814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b2801a7913aa27f445f51126156494463f19e814"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
