)]}'
{"id":"openstack%2Fskyline-console~901109","triplet_id":"openstack%2Fskyline-console~master~I4664429c0c01d195bed3701087e92776b068c69d","project":"openstack/skyline-console","branch":"master","topic":"default-project","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-11-17 04:26:22.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I4664429c0c01d195bed3701087e92776b068c69d","subject":"feat: support user\u0027s default project","status":"MERGED","created":"2023-11-16 06:43:48.000000000","updated":"2023-11-17 07:14:57.000000000","submitted":"2023-11-17 07:13:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901109-default-project","meta_rev_id":"56ce286de3721d4cdfc8de3d770f1dcec4a9dd37","_number":901109,"virtual_id_number":901109,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-17 07:13:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-11-17 02:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-11-17 02:29:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 09:57:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-17 02:29:45.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"fbc22251be64463580daf64536a5e7e92ce63967","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-16 06:43:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"721abeaebd7b8e4ae1e55fa2309c6573c2fa857e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-16 06:45:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6528b83deb0539601f035d2f48d53c9469756963","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-16 06:49:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fac8ddd8ff50e3bff28293f99f9c91ff67ebfe72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 09:57:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ca2982c6a4b41f3bb4efdbfb185fbb6\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c327c7803423444098ee351fa0ccc7f3 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3503c281ff54e9789ee82bd2e2fed02 : SUCCESS in 4m 24s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/b1e938b1f5664dacae343ae9c21a23ff : SUCCESS in 11m 03s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7e33ff21d9e94e3fbbf816e66aaa319c : SUCCESS in 4m 49s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/7df01bb320104d04a72fa13ede08ad85 : SUCCESS in 4m 47s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1622e69361e84a20924b3b9ad7c0f867 : TIMED_OUT in 3h 02m 37s","accounts_in_message":[],"_revision_number":3},{"id":"4dc4989ace1cce4b80e3026310a058e492411152","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-16 10:03:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"86abea1f00958485e2a02dea25b18ec39d95b49c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 13:13:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/287f03de39dd4d4a8c44aa874f768501\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8c8fc3d10c84dd28b2922f0812263b9 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01738e47840b45268f08e47e24e76cbc : SUCCESS in 4m 53s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0528bb387e8c456f9b55ffaedcead389 : SUCCESS in 11m 35s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/90c832c0625e4d5dad37a854e09d2752 : SUCCESS in 5m 57s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/562be5b54d0a4b1e8ba8d24253c931dd : SUCCESS in 8m 27s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/53e0ae533a89492d911437633202705f : TIMED_OUT in 3h 02m 49s","accounts_in_message":[],"_revision_number":3},{"id":"9c0e8d1512c6ddf6a84ea30ecd3d204a7ad6c864","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-17 01:11:08.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d6e2baa5713b7e4c4df3fe7d3275e41573795131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 02:20:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dc0161f5c8640968c4f7a57e7736f3c\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34962df2205a486d80af95ba2ec21299 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2142a3275bc4c4299998b9cba285d11 : SUCCESS in 4m 38s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/80c047d6d5f74c848fbeb379b0ae5393 : SUCCESS in 10m 06s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/56cc985d3411454a902b3127e5271220 : SUCCESS in 7m 18s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/852cb8ab6bec42c9b1c2c4b7b6d6dd4e : SUCCESS in 7m 13s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d71b8ea98d824c5c8f2faec5bb8b8674 : SUCCESS in 1h 04m 05s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ba9abb297abf4531abc5d529bac6d5a5 : SUCCESS in 1h 03m 44s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/aa1728373ea04d6fb4b7cb2d56ae93a4 : SUCCESS in 45m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9069bb66bc454a53817f21ebec1a3385 : SUCCESS in 41m 40s","accounts_in_message":[],"_revision_number":4},{"id":"0347d7aa4bfdf6a71b1326f152149360f44f3804","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-17 02:29:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d17a7165cb092ff363d81d0a9de520cd124b20e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 02:30:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7d9c8ae569969a1c2571d183db7e638c4e8b7bdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 04:14:36.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6d45a3c8d0e4386bff877d561dfeef3\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e95f26e993a04316b89cabc7c40337c4 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb32a281d2b466e98bae5b3a8781308 : SUCCESS in 5m 04s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c4fb958e2eda4bfcbea751a05f209faf : SUCCESS in 12m 55s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/c96f1cf58aa447908ffc228446c1bef4 : SUCCESS in 8m 06s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/873c4a2f746e4b3cb9444927f9ec12da : SUCCESS in 9m 05s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f72f85225cde415cb364699382923489 : SUCCESS in 1h 16m 19s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5c75c156dde84c15a4cb453aaf55aa11 : SUCCESS in 1h 18m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f688fa8f5d0f4d62959507bf29296a8d : POST_FAILURE in 31m 53s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3d5b1e59f6e84469b3d602e81e8dfb63 : SUCCESS in 1h 01m 26s","accounts_in_message":[],"_revision_number":4},{"id":"8fb7ae06ee720537c8027cde7898c4584a35cd75","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-17 04:26:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e719b51549ad43eacc997a28812a8fb9dcbe15c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 05:58:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0091770ff225420b9e7eaae9359ccd7c\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b4de7046c73442486734aebcd522130 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc6b7cfb81a4a81880c8897b614bd6b : SUCCESS in 4m 50s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/a6c92f99f13e407789fe7c0dbd0f13b4 : SUCCESS in 11m 35s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/26995d55a48140a0a74aa5fbb709c711 : SUCCESS in 7m 32s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e55b59cfd2324d3a8b88716cc628a146 : SUCCESS in 8m 12s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e4576fa1f1914b7baa25878e5ad46918 : SUCCESS in 1h 26m 41s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b2196d1e86964e93a961e60dc918722b : SUCCESS in 46m 42s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e0de4d17a7a94361ae7f0f671f6717dd : SUCCESS in 39m 50s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ad62eb0258874b32b3c7b42861cc7a5a : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":4},{"id":"112a655c187ed428c5b7bd242fc4ae5c97454f45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 05:59:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"11a31050f2dffa987feb10361fc2ff38acb3c104","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 07:13:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/114775929378406d8ff265f902713ce7\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d754ffe14010451182732191cd3ef441 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc123210f84d46f6a44a74303e89fff6 : SUCCESS in 4m 16s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4623ddfdf7d74289acf3c1e9342b1af3 : SUCCESS in 12m 10s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e5651dda91494d16995046420b42b3f6 : SUCCESS in 4m 41s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/888aff23505c4955bfce79b86a33564c : SUCCESS in 7m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e1f5b878ef834f93af578478e2323fee : SUCCESS in 1h 00m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/98b145f5470140e9958f358c7fd8486a : SUCCESS in 1h 09m 12s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7020b52bb4474537a537fedeb545dfa2 : SUCCESS in 42m 10s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b8f045ac24e04c7ba99c6c1c95d545ed : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":4},{"id":"628f1a2224462d42f128c3d1be806b67b256ddef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 07:13:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"56ce286de3721d4cdfc8de3d770f1dcec4a9dd37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 07:14:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/158c010ce7634be0b55a8d2b9e79dd31\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf5c220aaccf44b5bafc88f84d4c8953 : SUCCESS in 43s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5443d4fd75647a38bbfd9f7c667267e : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"968e2cf577cc4ded9544b5a495d6910ccb1133e7","revisions":{"ed1a422334f1151b0a863f01a45b3db496680b34":{"kind":"REWORK","_number":1,"created":"2023-11-16 06:43:48.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/09/901109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/901109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/1"}}},"commit":{"parents":[{"commit":"b2801a7913aa27f445f51126156494463f19e814","subject":"feat: update edit health monitor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b2801a7913aa27f445f51126156494463f19e814"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-16 06:39:29.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-16 06:43:11.000000000","tz":480},"subject":"feat: support user\u0027s default project","message":"feat: support user\u0027s default project\n\n1. support set user\u0027s default project\n2. display user\u0027s default project in the user list page/user detail page\n3. support remove user\u0027s default project\n\nChange-Id: I4664429c0c01d195bed3701087e92776b068c69d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ed1a422334f1151b0a863f01a45b3db496680b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ed1a422334f1151b0a863f01a45b3db496680b34"}]},"branch":"refs/heads/master"},"8e978b9eec385508f4eb1c531ab3a3371c68d08e":{"kind":"REWORK","_number":2,"created":"2023-11-16 06:45:17.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/09/901109/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/901109/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/2"}}},"commit":{"parents":[{"commit":"b2801a7913aa27f445f51126156494463f19e814","subject":"feat: update edit health monitor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b2801a7913aa27f445f51126156494463f19e814"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-16 06:39:29.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-16 06:45:04.000000000","tz":480},"subject":"feat: support user\u0027s default project","message":"feat: support user\u0027s default project\n\n1. support set user\u0027s default project\n2. display user\u0027s default project in the user list page/user detail page\n3. support remove user\u0027s default project\n\nChange-Id: I4664429c0c01d195bed3701087e92776b068c69d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e978b9eec385508f4eb1c531ab3a3371c68d08e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e978b9eec385508f4eb1c531ab3a3371c68d08e"}]},"branch":"refs/heads/master"},"306cddbba0dbc9717c9921d065b391a127676c47":{"kind":"REWORK","_number":3,"created":"2023-11-16 06:49:11.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/09/901109/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/901109/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/3"}}},"commit":{"parents":[{"commit":"b2801a7913aa27f445f51126156494463f19e814","subject":"feat: update edit health monitor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b2801a7913aa27f445f51126156494463f19e814"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-16 06:39:29.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-16 06:48:14.000000000","tz":480},"subject":"feat: support user\u0027s default project","message":"feat: support user\u0027s default project\n\n1. support set user\u0027s default project\n2. display user\u0027s default project in the user list page/user detail page\n3. support remove user\u0027s default project\n\nChange-Id: I4664429c0c01d195bed3701087e92776b068c69d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/306cddbba0dbc9717c9921d065b391a127676c47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/306cddbba0dbc9717c9921d065b391a127676c47"}]},"branch":"refs/heads/master"},"968e2cf577cc4ded9544b5a495d6910ccb1133e7":{"kind":"REWORK","_number":4,"created":"2023-11-17 01:11:08.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/09/901109/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/901109/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/901109/4"}}},"commit":{"parents":[{"commit":"892b511562420fa7d5e770b0e7ad831fa811e7c8","subject":"Merge \"feat: update extra specs of volume type and QoS spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/892b511562420fa7d5e770b0e7ad831fa811e7c8"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-16 06:39:29.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-17 01:09:59.000000000","tz":480},"subject":"feat: support user\u0027s default project","message":"feat: support user\u0027s default project\n\n1. support set user\u0027s default project\n2. display user\u0027s default project in the user list page/user detail page\n3. support remove user\u0027s default project\n4. update user e2e to adapt default project column\n\nChange-Id: I4664429c0c01d195bed3701087e92776b068c69d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/968e2cf577cc4ded9544b5a495d6910ccb1133e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/968e2cf577cc4ded9544b5a495d6910ccb1133e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
