)]}'
{"id":"openstack%2Fskyline-console~901116","triplet_id":"openstack%2Fskyline-console~master~Ibcf62cefb554172841ffee1fffad49cb2ba40dab","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-11-16 10:36:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibcf62cefb554172841ffee1fffad49cb2ba40dab","subject":"fix: Fix GPU type","status":"MERGED","created":"2023-11-16 07:23:45.000000000","updated":"2023-11-16 10:37:24.000000000","submitted":"2023-11-16 10:36:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901116","meta_rev_id":"61afd98d308f512e550f318fad8283edf4cdd5dd","_number":901116,"virtual_id_number":901116,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 10:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-11-16 09:28:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-11-16 09:28:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 08:47:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-16 09:28:02.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"190f2fa8c8f72a8a4d9c94027d67b9e1f81a8f36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2023-11-16 07:23:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"570b477089a15f48f83ca0fab2697430d7995f75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 08:47:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/308ea9000bce4a219be8dca6d07d49cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9df3b213f7348c4b0be0c3135e3b9e1 : SUCCESS in 5m 24s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8dca9e97032b45769f505834d0c0803c : SUCCESS in 9m 00s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/67a8c610a76646ed9fd4f7549f7a9595 : SUCCESS in 6m 44s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/cebd0e42f4a645588bca17a011e9c99a : SUCCESS in 5m 05s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d0188bd50ae8467cb2112a7bf424c3f5 : SUCCESS in 1h 23m 31s","accounts_in_message":[],"_revision_number":1},{"id":"644ef7aa773dfd7220494f91a16ad488f6523d8b","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-16 09:28:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d3e3e5311a8787efe2575bfe44f56fe4ceca7aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 09:28:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae38638b68050c12aec23200140b0373e41d9590","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 10:36:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9b19c79a24249989b5b061cf19863ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3986478e684f45cba94a06de4ff8ec04 : SUCCESS in 4m 59s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/29c3ee6af23944a8b60411ea709c7fb7 : SUCCESS in 10m 45s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/feb35c13085b4f76ac572a4705f8a5b4 : SUCCESS in 7m 40s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/fe9f9577be924d789e874daddd171b03 : SUCCESS in 4m 52s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/63ba04ece301426e8ad7ac017ca9f315 : SUCCESS in 1h 06m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6391471c7473cec94a7eefbf580372e86b8b8a84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 10:36:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"61afd98d308f512e550f318fad8283edf4cdd5dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 10:37:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1225b4157a74f998d4f7816dd037296\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf4c7a2218504e0098f1d8d3b5f5664a : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ccf2942c7863f0091491bef40deefd73be9b5dc5","revisions":{"ccf2942c7863f0091491bef40deefd73be9b5dc5":{"kind":"REWORK","_number":1,"created":"2023-11-16 07:23:45.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/16/901116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/16/901116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/901116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/901116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/901116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/16/901116/1"}}},"commit":{"parents":[{"commit":"b2801a7913aa27f445f51126156494463f19e814","subject":"feat: update edit health monitor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b2801a7913aa27f445f51126156494463f19e814"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-11-16 07:23:10.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2023-11-16 07:23:32.000000000","tz":480},"subject":"fix: Fix GPU type","message":"fix: Fix GPU type\n\nGPU type was lost for visual GPU\n\nChange-Id: Ibcf62cefb554172841ffee1fffad49cb2ba40dab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ccf2942c7863f0091491bef40deefd73be9b5dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ccf2942c7863f0091491bef40deefd73be9b5dc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
