)]}'
{"id":"openstack%2Fskyline-console~901219","triplet_id":"openstack%2Fskyline-console~master~I2fcbe8d79eb90250ed05509540bb2e29a310755d","project":"openstack/skyline-console","branch":"master","topic":"refresh","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-11-20 08:08:51.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I2fcbe8d79eb90250ed05509540bb2e29a310755d","subject":"feat: update refresh detail top without silence","status":"MERGED","created":"2023-11-17 08:01:42.000000000","updated":"2023-11-20 10:53:02.000000000","submitted":"2023-11-20 10:52:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901219-refresh","meta_rev_id":"0782250eac32a9b876fface7ffae3869a27da554","_number":901219,"virtual_id_number":901219,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-20 10:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-11-17 09:28:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-11-17 09:28:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-17 09:17:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-17 09:28:39.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"4c3256f6d79215b9a2d27777cc7d680843233f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-17 08:01:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"caa82c646343f80f715ad89d01e13a62af053ffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 09:17:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b8451321891493b8c2e247b0739bd77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af6584b303764fc3982cb2e77a2b6465 : SUCCESS in 4m 35s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/1281a9fd5d484380bb64348a3efc504c : SUCCESS in 7m 28s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6eaae5e7629549a58b2fc201dff019c2 : SUCCESS in 7m 25s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/9c6c484de11d4f63b2aee77e27f26f92 : SUCCESS in 6m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/67b29a67184e4343ae34f5f973395990 : SUCCESS in 1h 06m 27s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1885d89fe29b4601aae74ac10b60dc7c : SUCCESS in 1h 10m 08s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/42e42126ee574619af350ff36684f163 : SUCCESS in 58m 31s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2c4d94ce7bcb4a0f8f0438d0c259f1f0 : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":1},{"id":"a17d4b3bfb534e371c38416bfb5bf9f9694408be","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-17 09:28:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"705bdf52063fa8db7347605ccf89bc008487da35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 09:29:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b5b6ec42316c62321755589d69e77c26d322c87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 10:45:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/91ee68514b3447d7bf417e27fb7e677d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1870ca7eb8c402c9eacc60d8921e239 : SUCCESS in 4m 34s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8f9a8659a47440eebfd1ed8732958c49 : SUCCESS in 8m 27s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/f47f5db4efc34102aaf990f027aebc80 : SUCCESS in 10m 18s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0b093f4e35b0481daba045d65f3d75d5 : SUCCESS in 5m 42s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/155054c70b524b658e0fa290b63a9f74 : SUCCESS in 1h 15m 40s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9ced8d0055d84fd599603101d18acda7 : SUCCESS in 1h 01m 59s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bc31dd8aedb34d8e816fdedbcbad805d : FAILURE in 1h 02m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/76c7eada7d674e68928100a77ca4595d : SUCCESS in 51m 58s","accounts_in_message":[],"_revision_number":1},{"id":"346e30493891cc70ffdcee91d6bed993a8f70f82","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-20 02:41:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"370c1194cc8e433c4167a60ba92d40053bd507fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 04:15:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96ba431205b6435c82f399721869e67d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85102a2b874a4e2f95eb41a6c0d1f499 : SUCCESS in 4m 28s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ccb4bf235c764f4c85f4c3733e702466 : SUCCESS in 10m 35s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e9603db14efb4ce3b511456035643908 : SUCCESS in 8m 18s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0c8485c103504847854ee9560c40a130 : SUCCESS in 7m 50s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fdf3020fbd1b40318e3095ee7f346e72 : SUCCESS in 1h 21m 08s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c3c82c54fbe74a708ed74b9c00e9d5d8 : SUCCESS in 1h 04m 35s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/4977d03f18fd4bc6bcd053a6d4c08ae8 : SUCCESS in 1h 00m 30s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a69c49bb55f4442a9d779005e048f146 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c95ce4c6c74dd1f83f73054846c0033516135e87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 04:15:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf02ce913776f3b38de8d757102d65e7b05e8890","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 05:38:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c27ca2b9391546259af60de36a616fcb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c77f89dacf4aa1838711b355f31260 : SUCCESS in 4m 02s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ef76fb66953e4278b86a6e0d3b1dcd42 : SUCCESS in 10m 54s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7cb492a6a682436b9cd4054edba5012b : SUCCESS in 4m 56s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/1ef14f9041674b35bec3577adb9db912 : SUCCESS in 6m 40s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/572ff01df28e4ba8a8326bdbb9f63e4f : FAILURE in 1h 08m 30s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9903df7f640d4490b83a50721fafd483 : FAILURE in 49m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/df12967268d246ceba2920f5ed8eef04 : FAILURE in 28m 12s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f08aa8c1663d460e9b5afc6b9f4c8dea : FAILURE in 28m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b6c2b05656d31877648ae289dea0faaffe3bb9b7","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-20 06:55:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47a3cca7675d270027fecc368c7e6f2587d9ab6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 08:06:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce805c1cbdf74df1be77d3c6dc111bb9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fc2f7b4d6dd4d599d6b5c8026ef2f31 : SUCCESS in 4m 57s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ddc9add3ed9b4e978a1d970b249a2a7a : FAILURE in 3m 50s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0d98e7db647c40e59180cd430d3c767b : SUCCESS in 5m 11s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4cf19923672c4431b0a28c9ddbfa921f : SUCCESS in 5m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/baf3973920294a04b85912c9d515e876 : SUCCESS in 1h 09m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a580d3f4a7334b5184f5cfaf54bd155c : SUCCESS in 1h 04m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6a171ed6bdd145f5860ca7c904f45a56 : SUCCESS in 46m 49s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2d73fab340934bd49f8e9133cd0f696c : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0eddd26c87899b6822d2e0bfcb4b50951e0d9808","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-20 08:08:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebb667010c69677322ff11cfea760f20b0f11d53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 09:14:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92b9399763604653a314a77df9acca9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16501b3b7f1f433ab9ba9a087cca0671 : SUCCESS in 3m 59s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/1ecaaeb367e641d1a7a26532ecce5cf4 : SUCCESS in 6m 42s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/20516aea00784b59bbfcb4ab821bca4a : SUCCESS in 7m 40s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/fc61abb96cb84792b7653a5f4eef8e98 : SUCCESS in 7m 39s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6400d13ab3ed423eb21d51e0f9af60ff : SUCCESS in 1h 04m 42s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/513a9a4cf0a0416db368a2db2603b50f : SUCCESS in 51m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b63f7968385542629cad2f8559487b55 : SUCCESS in 45m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/92ed7f2be802430b8dc49abfa19a4509 : SUCCESS in 44m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d65a975a27e6daba3bc096e06743362b1c1de367","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 09:15:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34ae80f90f17fb3de979a5a2a45f137fe59e6674","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 10:52:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d71215fafaf64755888e828bbc181eba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee1e801e7f1649118ca097070164e18f : SUCCESS in 4m 31s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0239a61d660345b58c8500ad75d9131a : SUCCESS in 6m 39s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7d3e0f456feb4677a390f50df1e5f269 : SUCCESS in 4m 35s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d86d6204f7e545488013b1c583cce863 : SUCCESS in 7m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fb92d9b987234f7699c0093822460689 : SUCCESS in 1h 29m 19s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c2ca89a600c040b9922b360bb00cfe98 : SUCCESS in 47m 42s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a39d3e6929db4259911f7a90425fa403 : SUCCESS in 58m 18s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/34e43bdb9bd041d195c36b060e7f4eb2 : SUCCESS in 38m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6df266bfff9fa51548b514eb8373f3392fec2a52","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 10:52:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0782250eac32a9b876fface7ffae3869a27da554","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 10:53:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13f9881c942b475a8fa706579b878920\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a20ad4b25541a78b2dcb5223513de0 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c356e214784a79f988d785021f6bf901aa34601b","revisions":{"c356e214784a79f988d785021f6bf901aa34601b":{"kind":"REWORK","_number":1,"created":"2023-11-17 08:01:42.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/19/901219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/19/901219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/19/901219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/19/901219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/19/901219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/19/901219/1"}}},"commit":{"parents":[{"commit":"968e2cf577cc4ded9544b5a495d6910ccb1133e7","subject":"feat: support user\u0027s default project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/968e2cf577cc4ded9544b5a495d6910ccb1133e7"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-17 08:00:43.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-17 08:00:43.000000000","tz":480},"subject":"feat: update refresh detail top without silence","message":"feat: update refresh detail top without silence\n\n1. For some list tab in the detail page, when submit actions in the list tab, the detail info and the list tab is better to refresh without silence.\n2. Refresh the whole user detail page when the set/remove default project of the user\n\nChange-Id: I2fcbe8d79eb90250ed05509540bb2e29a310755d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c356e214784a79f988d785021f6bf901aa34601b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c356e214784a79f988d785021f6bf901aa34601b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
