)]}'
{"id":"openstack%2Fskyline-console~901228","triplet_id":"openstack%2Fskyline-console~master~I50e7d1882c83563dd303a43951f08e288f991c17","project":"openstack/skyline-console","branch":"master","topic":"qos","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-11-18 10:45:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50e7d1882c83563dd303a43951f08e288f991c17","subject":"fix: update the search label in the extra spec","status":"MERGED","created":"2023-11-17 09:40:46.000000000","updated":"2023-11-18 10:47:36.000000000","submitted":"2023-11-18 10:45:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901228-qos","meta_rev_id":"e6c9081654bc7fb81d00a93a1887a53e0befa1c2","_number":901228,"virtual_id_number":901228,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-18 10:45:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-11-18 09:08:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-11-18 09:08:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-17 10:27:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-18 09:08:56.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"a7b147f32808a4e4e6e52a5b7d7bbf5f243799ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-17 09:40:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d70f8781216c55c5780258e4f7cd7f702c0a1a5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 10:27:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fff65a163bb142ecb44358908deb4b43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/163a3abca5664192ba98388282d2b66f : SUCCESS in 4m 42s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/b4050a7648354baa948bae3b560f0c95 : SUCCESS in 7m 40s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ba45e0e1f21841eeaae1f18808ce4cba : SUCCESS in 7m 54s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/20d0077756814be8afb6ca550d614fdd : SUCCESS in 6m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/55eb7f3f4ad64dad8536a635e0f21594 : SUCCESS in 45m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0b89edfe71b6fbb5038ea2ad151090c4c432bde2","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-18 09:08:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81796e55519d9b24ae28303d219655879af45be7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-18 09:09:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"770db1231402680e7ddac319d462bbd57d995a55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-18 10:45:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe0dafa9e06a409aaf37e99cda370877\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8bf65238fd846f7bb29c0b36e974f5d : SUCCESS in 4m 31s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9efd343617fb4f3a87f328402cb54f06 : SUCCESS in 11m 07s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0e219fffdff9476a999c023ad434c0fc : SUCCESS in 7m 38s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/527787789c2248c3a9ef6e7918d38305 : SUCCESS in 7m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c25c00550a654b7c89beef11634ed674 : SUCCESS in 42m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7c1d6bb4f397c6e7c1a136f6955902540137de32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-18 10:45:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e6c9081654bc7fb81d00a93a1887a53e0befa1c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-18 10:47:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26957ea5b2c845cd8b8e5c14cef24233\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35142037a305484088bb6ffa24efff3a : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21ac632a43ec22d3ceeac92bf457c3539861d3b9","revisions":{"21ac632a43ec22d3ceeac92bf457c3539861d3b9":{"kind":"REWORK","_number":1,"created":"2023-11-17 09:40:46.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/28/901228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/901228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/901228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/901228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/901228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/901228/1"}}},"commit":{"parents":[{"commit":"968e2cf577cc4ded9544b5a495d6910ccb1133e7","subject":"feat: support user\u0027s default project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/968e2cf577cc4ded9544b5a495d6910ccb1133e7"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-17 09:40:22.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-17 09:40:22.000000000","tz":480},"subject":"fix: update the search label in the extra spec","message":"fix: update the search label in the extra spec\n\n1. update the search label in the extra spec tab in the volume type detail page\n2. update the search label in the extra spec tab in the Qos spec detail page\n\nChange-Id: I50e7d1882c83563dd303a43951f08e288f991c17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/21ac632a43ec22d3ceeac92bf457c3539861d3b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/21ac632a43ec22d3ceeac92bf457c3539861d3b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
