)]}'
{"id":"openstack%2Fskyline-console~901448","triplet_id":"openstack%2Fskyline-console~master~Ib7e7da84de8e884e2cec65fa9747f04551e2de96","project":"openstack/skyline-console","branch":"master","topic":"DeleteVolumeInstance","attention_set":{},"removed_from_attention_set":{"20865":{"account":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"last_update":"2023-11-22 07:32:08.000000000","reason":"Change was submitted"},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2023-11-22 07:32:08.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-11-22 05:48:28.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"36553":{"account":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"last_update":"2023-11-22 07:32:08.000000000","reason":"Change was submitted"},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2023-11-22 07:32:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib7e7da84de8e884e2cec65fa9747f04551e2de96","subject":"feat: Allow deletion of Volume on Instance Delete setting","status":"MERGED","created":"2023-11-20 08:58:01.000000000","updated":"2023-11-22 07:33:13.000000000","submitted":"2023-11-22 07:32:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901448-DeleteVolumeInstance","meta_rev_id":"3116540cd02ab065e58046030b9e8014139ecdea","_number":901448,"virtual_id_number":901448,"owner":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-22 07:32:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2023-11-22 05:48:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2023-11-22 05:48:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}],"reviewers":{"REVIEWER":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-20 09:19:42.000000000","updated_by":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2023-11-20 09:19:42.000000000","updated_by":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2023-11-20 09:19:42.000000000","updated_by":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"REVIEWER"},{"updated":"2023-11-20 09:19:42.000000000","updated_by":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2023-11-20 10:11:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92b55ad604b273619edaa08e90c3504684bf9131","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"date":"2023-11-20 08:58:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cacae9a28c3152f684bd22b1681073ab6ffcde1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 10:11:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1d0f9d723674dbfa6c9cd6e8fb98739\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfeaf0431f81457595d6016d422e4e6f : SUCCESS in 5m 20s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/09610e4204954a6da2bc278337692fbb : SUCCESS in 11m 09s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1d015e5cf33043f89af1cad7d782aec4 : SUCCESS in 7m 11s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/840dce206c1341278fdac3052fa83360 : SUCCESS in 7m 46s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/cd7e509495e34428a1e0c8030127e78c : SUCCESS in 1h 11m 37s","accounts_in_message":[],"_revision_number":1},{"id":"da7d6656de945c5606da0c8ea4cf59805e2fcb66","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-21 01:44:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fd088d789d6885e1afb8c5d6170c6250f0eff0be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"date":"2023-11-21 04:49:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b6a40f9912c15bfed2b85e05c0ca6f201336119f","author":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"date":"2023-11-21 04:53:45.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"85699234a6871313de9c1534b41c64f47317ce6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 05:52:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/952249f338cb431c9abcdaa98b9e653b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c019286dbc84535a6aa8d363fe4ea48 : SUCCESS in 4m 36s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/979e884cbfc74d948da1d61649ac2929 : SUCCESS in 6m 32s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1cbf01da6b3f4151bb898b6798cae5ee : SUCCESS in 4m 30s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2e4dbddb2ea64edcb1e3f33f1601c03f : SUCCESS in 5m 28s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2c4610e377c442a5a0430acb93ebd94f : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":2},{"id":"795a9362a69bc904c836093a814dd2c7b14d5236","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-21 05:56:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"085fb1615daf74f692741fb8d4c63c492ca294c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"date":"2023-11-21 11:19:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3ebe300e6656b6fe3eb983525bb89c4e74be1b5f","author":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"date":"2023-11-21 11:27:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b1134fc2356719fdafd36f376fa8a9dd28bb2d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 12:58:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/279900f8a2dd465aa59253e1440b9d4d\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/922633b9f758466b999ec5bf508f9e45 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05ff417768c04b8fa2f11025937b6625 : SUCCESS in 4m 29s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f9956420f0194c8fb8b9d60e563b69bf : SUCCESS in 9m 28s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4ae703112ccc4d838de9d9979fd89619 : SUCCESS in 5m 13s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/24890845cdbd4fb390e3683ea5a423d8 : SUCCESS in 7m 11s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/84077d5d09294dd1810df375329de5d7 : SUCCESS in 1h 32m 25s","accounts_in_message":[],"_revision_number":3},{"id":"dbcc7eebef8ce5513b3f6095e315de7ea67f3904","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-22 02:11:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f5894fa419a199697a9f40270298ff9c3c5de78c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"date":"2023-11-22 04:14:16.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"72d61694b0f560652f0e35e3a701ff4f4a5aa0b9","author":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"date":"2023-11-22 04:15:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"48af07d9814513fc0319e9133667c16f41cf0591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 05:45:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1762bdc926af4249a66693504449a109\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/803266be401641a1857bda7d6c8e71c7 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/324878ea019a4b38b9a7b5e01730383e : SUCCESS in 4m 30s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9cf4e2c419244e13afefff5fa92e78f5 : SUCCESS in 6m 38s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e2f443feba1641c0a7fe125360c862af : SUCCESS in 4m 51s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e41537acbc8f48f1bada9860d367359b : SUCCESS in 11m 14s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/72a9f815353d4a7ba505caa580c9fc11 : SUCCESS in 1h 25m 32s","accounts_in_message":[],"_revision_number":4},{"id":"82e829563ae4c7fd6843983354634938e6fc9891","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-22 05:48:28.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bbd2e10c70267c9c5fcac53f8028fe18bafb63ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 05:49:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b542068e193783823388b6bb2af3710211f15bfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 07:32:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48d1e17f3fea494b96f529ea5872030c\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3ae548d60024931adfc2ca17cc12d9e : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baeb3c62faf8445893b7d063f05893e2 : SUCCESS in 4m 01s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/7b77aab3599843fd9571770b67aedda2 : SUCCESS in 7m 51s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/dcdc1b7959f740bb91a0a420572e7169 : SUCCESS in 7m 34s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/26bde315dfad4d7cbe81e9930aa2531d : SUCCESS in 5m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ed22eb1b886d41c08c1547d798ec5430 : SUCCESS in 1h 35m 50s","accounts_in_message":[],"_revision_number":4},{"id":"b48946ec48860d915af6a0fa58a7fcd2c317abed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 07:32:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3116540cd02ab065e58046030b9e8014139ecdea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 07:33:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d47fef2049f14cbab1adbe30e4c2451c\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23e21aa87fe94709b00ee2fee01f1c99 : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc9d49df2ed04a4d92b99527a5a24d5a : SUCCESS in 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"639bcedd7d3e93808aef93c84a87c32378205a3f","revisions":{"a968e994ac08d28918140f7f4d3b2fe891cd9977":{"kind":"REWORK","_number":1,"created":"2023-11-20 08:58:01.000000000","uploader":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"ref":"refs/changes/48/901448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/901448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/1"}}},"commit":{"parents":[{"commit":"23bc6f8e2195eab104304341d394ec1bda5c92e6","subject":"Merge \"fix: update the search label in the extra spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/23bc6f8e2195eab104304341d394ec1bda5c92e6"}]}],"author":{"name":"Elly Murillo","email":"elly.murillo@qumulus.io","date":"2023-11-20 08:57:16.000000000","tz":480},"committer":{"name":"Elly Murillo","email":"elly.murillo@qumulus.io","date":"2023-11-20 08:57:16.000000000","tz":480},"subject":"feat: Allow deletion of Volume on Instance Delete setting","message":"feat: Allow deletion of Volume on Instance Delete setting\n\nThis feature allows the user to check whether the selected Volume should be deleted along with the instance\nwhen the user selected the Bootable Volume Start Source via Instance creation module.\n\nChange-Id: Ib7e7da84de8e884e2cec65fa9747f04551e2de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a968e994ac08d28918140f7f4d3b2fe891cd9977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a968e994ac08d28918140f7f4d3b2fe891cd9977"}]},"branch":"refs/heads/master"},"b6df2541f1136a2a358a4a0e7105dee7c395fed5":{"kind":"REWORK","_number":2,"created":"2023-11-21 04:49:11.000000000","uploader":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"ref":"refs/changes/48/901448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/901448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/2"}}},"commit":{"parents":[{"commit":"23bc6f8e2195eab104304341d394ec1bda5c92e6","subject":"Merge \"fix: update the search label in the extra spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/23bc6f8e2195eab104304341d394ec1bda5c92e6"}]}],"author":{"name":"Elly Murillo","email":"elly.murillo@qumulus.io","date":"2023-11-20 08:57:16.000000000","tz":480},"committer":{"name":"Elly Murillo","email":"elly.murillo@qumulus.io","date":"2023-11-21 04:48:19.000000000","tz":480},"subject":"feat: Allow deletion of Volume on Instance Delete setting","message":"feat: Allow deletion of Volume on Instance Delete setting\n\nThis feature allows the user to check whether the selected Volume should be deleted along with the instance\nwhen the user selected the Bootable Volume Start Source via Instance creation module.\n\nChange-Id: Ib7e7da84de8e884e2cec65fa9747f04551e2de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b6df2541f1136a2a358a4a0e7105dee7c395fed5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b6df2541f1136a2a358a4a0e7105dee7c395fed5"}]},"branch":"refs/heads/master"},"ee1aa289190bb49b080fc7625d1e95f6f0b6c98e":{"kind":"REWORK","_number":3,"created":"2023-11-21 11:19:01.000000000","uploader":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"ref":"refs/changes/48/901448/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/901448/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/3"}}},"commit":{"parents":[{"commit":"23bc6f8e2195eab104304341d394ec1bda5c92e6","subject":"Merge \"fix: update the search label in the extra spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/23bc6f8e2195eab104304341d394ec1bda5c92e6"}]}],"author":{"name":"Elly Murillo","email":"elly.murillo@qumulus.io","date":"2023-11-20 08:57:16.000000000","tz":480},"committer":{"name":"Elly Murillo","email":"elly.murillo@qumulus.io","date":"2023-11-21 11:18:19.000000000","tz":480},"subject":"feat: Allow deletion of Volume on Instance Delete setting","message":"feat: Allow deletion of Volume on Instance Delete setting\n\nThis feature allows the user to check whether the selected Volume should be deleted along with the instance\nwhen the user selected the Bootable Volume Start Source via Instance creation module.\n\nChange-Id: Ib7e7da84de8e884e2cec65fa9747f04551e2de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ee1aa289190bb49b080fc7625d1e95f6f0b6c98e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ee1aa289190bb49b080fc7625d1e95f6f0b6c98e"}]},"branch":"refs/heads/master"},"639bcedd7d3e93808aef93c84a87c32378205a3f":{"kind":"REWORK","_number":4,"created":"2023-11-22 04:14:16.000000000","uploader":{"_account_id":36553,"name":"Elly Murillo","display_name":"Elly","email":"elly.murillo@qumulus.io","username":"ellymurillo"},"ref":"refs/changes/48/901448/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/901448/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/901448/4"}}},"commit":{"parents":[{"commit":"23bc6f8e2195eab104304341d394ec1bda5c92e6","subject":"Merge \"fix: update the search label in the extra spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/23bc6f8e2195eab104304341d394ec1bda5c92e6"}]}],"author":{"name":"Elly Murillo","email":"elly.murillo@qumulus.io","date":"2023-11-20 08:57:16.000000000","tz":480},"committer":{"name":"Elly Murillo","email":"elly.murillo@qumulus.io","date":"2023-11-22 04:13:58.000000000","tz":480},"subject":"feat: Allow deletion of Volume on Instance Delete setting","message":"feat: Allow deletion of Volume on Instance Delete setting\n\nThis feature allows the user to check whether the selected Volume should be deleted along with the instance\nwhen the user selected the Bootable Volume Start Source via Instance creation module.\n\nChange-Id: Ib7e7da84de8e884e2cec65fa9747f04551e2de96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/639bcedd7d3e93808aef93c84a87c32378205a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/639bcedd7d3e93808aef93c84a87c32378205a3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
