)]}'
{"id":"openstack%2Fskyline-console~901698","triplet_id":"openstack%2Fskyline-console~master~I2078ac1b18914cf39e1254c65788cdd53bac0410","project":"openstack/skyline-console","branch":"master","topic":"bootable-volume","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-11-23 09:17:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2078ac1b18914cf39e1254c65788cdd53bac0410","subject":"fix: fix i18n for deleteVolumeInstance formItem label","status":"MERGED","created":"2023-11-23 06:06:44.000000000","updated":"2023-11-23 09:19:00.000000000","submitted":"2023-11-23 09:17:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901698-bootable-volume","meta_rev_id":"a9f1096d43a72c9e4066b08324ab38705d6648e6","_number":901698,"virtual_id_number":901698,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-23 09:17:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-11-23 08:07:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-11-23 08:07:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-23 07:14:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-23 08:07:26.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"d3bfb6105174dfe36c1fb209b7e0adc7db730c68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-23 06:06:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4437535ff06812a322e44da58181f0fe825cc429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 07:14:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c084e818bf4a4f9d8545a4c95ac986a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f96a8316ea4f4a098bb8fc4338f99ea0 : SUCCESS in 5m 13s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ddc1acd26a6945f1a7fc8fc1ae4f6aa0 : SUCCESS in 9m 48s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/27e61efeda984c819b8cf6bf9c7d7f2c : SUCCESS in 6m 12s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/c581738912934af7ba71baf99b14d426 : SUCCESS in 7m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/434c8f43aa94404987062ee7641ba639 : SUCCESS in 1h 06m 46s","accounts_in_message":[],"_revision_number":1},{"id":"006fdf1f843c76e79b37246ac4bd651fe0d7e9d9","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-23 08:07:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"744b612ad80f8b7e18dd0c53303800aab6f35290","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 08:07:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9a5830a0a70a43882c584d1af53f32bd1b475fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 09:17:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5292addce204b50ae4b30b3a89d2c34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/065e31792ce84ad29236adbc343bdfab : SUCCESS in 4m 17s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/391bb56a7aa64510b07035581abe410d : SUCCESS in 10m 16s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/51f2e93db1d245f59b2235d4b7b79f7b : SUCCESS in 5m 21s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f6c97c61d64c4fffb35a97308e984dda : SUCCESS in 5m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c3202d8c019f4f2ea56c00ef649c714f : SUCCESS in 1h 09m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0aa421dcc9ab70588fbe982e1eb4d8082486c275","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 09:17:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a9f1096d43a72c9e4066b08324ab38705d6648e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 09:19:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25b3017cf50f487596c0aefd995e02dd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44f7d56732094c3eba874aab8524cfbc : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6cd3d83eb0c99198ed311b994c54728b9ad5019","revisions":{"e6cd3d83eb0c99198ed311b994c54728b9ad5019":{"kind":"REWORK","_number":1,"created":"2023-11-23 06:06:44.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/98/901698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/98/901698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/98/901698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/98/901698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/98/901698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/98/901698/1"}}},"commit":{"parents":[{"commit":"54e64ea9fc25045952a9fbfdfbc1689e0c3b5d5a","subject":"fix: Fix action colum of the Table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/54e64ea9fc25045952a9fbfdfbc1689e0c3b5d5a"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-23 06:06:16.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-23 06:06:16.000000000","tz":480},"subject":"fix: fix i18n for deleteVolumeInstance formItem label","message":"fix: fix i18n for deleteVolumeInstance formItem label\n\nFix the label of the deleteVolumeInstance formItem\n\nChange-Id: I2078ac1b18914cf39e1254c65788cdd53bac0410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e6cd3d83eb0c99198ed311b994c54728b9ad5019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e6cd3d83eb0c99198ed311b994c54728b9ad5019"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
