)]}'
{"id":"openstack%2Fskyline-console~901807","triplet_id":"openstack%2Fskyline-console~master~Ifdb16cc3b82dbb928f4c2e3fcf336450cbe7cc3c","project":"openstack/skyline-console","branch":"master","topic":"lb","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2023-11-24 03:58:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifdb16cc3b82dbb928f4c2e3fcf336450cbe7cc3c","subject":"feat: support admin_state_up for listener\u0027s pool","status":"MERGED","created":"2023-11-24 01:19:04.000000000","updated":"2023-11-24 04:00:48.000000000","submitted":"2023-11-24 03:58:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901807-lb","meta_rev_id":"1ace1e0d99073ffab1c1a49726cb9d75a47643c1","_number":901807,"virtual_id_number":901807,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-24 03:58:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2023-11-24 02:40:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2023-11-24 02:40:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-24 02:36:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-24 02:40:57.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"96c1a5d3de9393fc445728ad7d4d46cb64a8fdfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2023-11-24 01:19:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa38915929552e023b44630e7ed044bfc5a92358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 02:36:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06446fe4ba0547a484cfac375c2bf58d\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd2bb2fb92664a49bf59fde56b40fa07 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/570c873d7813467fb2c15e64751c5957 : SUCCESS in 5m 40s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/607e04a27af44da9aa490cefa55144ee : SUCCESS in 8m 26s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b0fd58408f494f86b58516fd23d768e4 : SUCCESS in 7m 58s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/42c6115d0d14428f8d05f966aa0ab51c : SUCCESS in 7m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9d9ac988ea4049549f232089d1ea24d7 : SUCCESS in 1h 09m 11s","accounts_in_message":[],"_revision_number":1},{"id":"289c44e3cee23ade41f9c3d0cb4613500c716161","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2023-11-24 02:40:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4895e26115982f1bac0703828a1dad817ac80e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 02:41:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4faa8ead0d3dba6a387a5e1c2d526a9b4efafcba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 03:58:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e01c5d463b584d94896bffc1a768e04a\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c9aeccec512489ca30ad300fe566e10 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ceb709742fa40a9b696afdf6a9bb793 : SUCCESS in 5m 13s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/7014aa3e34f343f18736c38477798ab5 : SUCCESS in 12m 51s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/58161d99b64247c6899ff9d9a83f56a5 : SUCCESS in 8m 30s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b34bfb84daa24eff9967892ebaabcef3 : SUCCESS in 8m 31s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/964c89dc80c74f3c932543e2948ad821 : SUCCESS in 1h 10m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4601ce6c6092997b2ff22f6b251ce49615a24e5e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 03:58:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1ace1e0d99073ffab1c1a49726cb9d75a47643c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 04:00:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75f5b38c390d4eca8a36ad0666940de7\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05b4e42be0fc421a8136688ae337291d : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02bad326733c4b31809f4bfa0683aa15 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44bf15d7e8504152172f092eac3a4318aafdc26d","revisions":{"44bf15d7e8504152172f092eac3a4318aafdc26d":{"kind":"REWORK","_number":1,"created":"2023-11-24 01:19:04.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/07/901807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/901807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/901807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/901807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/901807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/901807/1"}}},"commit":{"parents":[{"commit":"9fcb6b34a8a819ecbd54d38408e71ecd0e1866a4","subject":"fix: fix remove insert_headers for listener","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9fcb6b34a8a819ecbd54d38408e71ecd0e1866a4"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-24 01:18:39.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-24 01:18:39.000000000","tz":480},"subject":"feat: support admin_state_up for listener\u0027s pool","message":"feat: support admin_state_up for listener\u0027s pool\n\n1. support setting admin_state_up when creating listener\u0027s pool in the listener detail page\n2. update pool and health monitor cards in the listener detail page\n\nChange-Id: Ifdb16cc3b82dbb928f4c2e3fcf336450cbe7cc3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/44bf15d7e8504152172f092eac3a4318aafdc26d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/44bf15d7e8504152172f092eac3a4318aafdc26d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
