)]}'
{"id":"openstack%2Fskyline-console~903570","triplet_id":"openstack%2Fskyline-console~master~Ib985256ec8d8e6d12dcee147b2324271b233ad34","project":"openstack/skyline-console","branch":"master","topic":"i18n-korean-translation","attention_set":{},"removed_from_attention_set":{"36208":{"account":{"_account_id":36208,"name":"Kim HyoBin","display_name":"HyoBin","email":"hb.kim@okestro.com","username":"hbkim"},"last_update":"2024-01-17 07:16:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib985256ec8d8e6d12dcee147b2324271b233ad34","subject":"korean localization translation HMS DMY format","status":"MERGED","created":"2023-12-13 02:09:18.000000000","updated":"2024-01-17 07:18:00.000000000","submitted":"2024-01-17 07:16:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903570-i18n-korean-translation","meta_rev_id":"647a71c13d73e60efa6ee3576cb04198b3f3921e","_number":903570,"virtual_id_number":903570,"owner":{"_account_id":36208,"name":"Kim HyoBin","display_name":"HyoBin","email":"hb.kim@okestro.com","username":"hbkim"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-17 07:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2024-01-17 03:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2024-01-17 05:39:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-16 15:26:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2024-01-17 05:39:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-13 04:56:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-14 02:50:26.000000000","updated_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"reviewer":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"state":"CC"},{"updated":"2024-01-16 15:26:30.000000000","updated_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"reviewer":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"state":"REVIEWER"},{"updated":"2024-01-17 03:01:13.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2024-01-17 05:39:04.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"44cc103eff49d8e65e0c16c3e43b481b592d77be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36208,"name":"Kim HyoBin","display_name":"HyoBin","email":"hb.kim@okestro.com","username":"hbkim"},"date":"2023-12-13 02:09:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eedaa0e0a9e90e515e79dd510d63e5345cb856e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 04:56:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/55553dd5e1524ed38280eb0b97422339\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5976bb32e9cb4be8977c1b001baa79b2 : SUCCESS in 5m 46s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8374a41037bb403db7e16dcd81780c3c : SUCCESS in 9m 27s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/82b17ab8edd547c0b67b0126e7e516b6 : SUCCESS in 7m 01s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/3a571e363218482ab29cef2f8c492a16 : SUCCESS in 8m 19s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9762968ea7e946e09b9aad2e735e0382 : SUCCESS in 1h 35m 11s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5826f0cdf0304750819d9471b8746c14 : SUCCESS in 1h 05m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9ccb4a50d4df49059c63486e9d18eca4 : FAILURE in 56m 36s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3cb4002be6f742758b122110999ab518 : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b9476bcdc0b1a3f433a11c6c21ab06877d07eac1","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2023-12-14 02:50:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ce49fdddc98f90eab980e0a4b04c5695365356a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 04:50:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c10ef9e8eeca4cc9ae1d345c91bc4542\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f766a5c44c074a67899549e79eb1666c : SUCCESS in 4m 43s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d15d264f654941bfb5b4739063061d32 : SUCCESS in 10m 28s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/8e3235b2de22479295e564e792c39a1e : SUCCESS in 5m 23s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b4954a1392bb4dbebb83b6d1e1aa3368 : SUCCESS in 8m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ce01d63869d74360b10721ea649b5363 : SUCCESS in 1h 24m 00s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5d4b03ee759a40b48094ea5378ddb6fe : SUCCESS in 1h 04m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/30bc07e87ca448ea94f5eab5283c6bd4 : SUCCESS in 46m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/be29bba73c6c43f880950673974622c1 : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":1},{"id":"bafd071deae1c1c454bdc9561f2935355a1fd924","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2024-01-16 15:26:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"49b48ea0616f20be53aadc5989896a198abf9df3","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-01-17 03:01:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"381ca1378d5d28b237442af324b63d6cbde6a6a9","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-01-17 05:39:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc53e5cd0b383d9673bec82067193d09cbfc6fbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 05:39:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"998a903c72c7dd6f9cb69b53852c0a0f7229cbf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 07:16:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85b1f697bc844a0cb46715ea47b0fdaf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab15be9de6284c95b6a678c230ae6cca : SUCCESS in 5m 24s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/bda10ddacc5d4193baaf6646cccffa31 : SUCCESS in 7m 18s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/2b05448373654c5ba4c1c6fd16d60350 : SUCCESS in 7m 24s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/7673ba432519440ab82ec9ebd8abaf8e : SUCCESS in 5m 33s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ecaa1f4310bc44fa824cb279ad4765bd : SUCCESS in 1h 19m 47s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/69ec77285d3f497ab2a1f64d0e195360 : SUCCESS in 1h 08m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f101f23616964d7b8a9b624d514523dd : SUCCESS in 41m 50s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d4d21350d7e44b65960cee1a6a558212 : SUCCESS in 58m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c1938eb9a802599f278c004a56e4618134ff57db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 07:16:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"647a71c13d73e60efa6ee3576cb04198b3f3921e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 07:18:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88001d912e9041e39cd6b22aef9c159d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6079a22d545408aaaac08570a8ef2b9 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6480faa148270571bdca3448c17455376d74f082","revisions":{"6480faa148270571bdca3448c17455376d74f082":{"kind":"REWORK","_number":1,"created":"2023-12-13 02:09:18.000000000","uploader":{"_account_id":36208,"name":"Kim HyoBin","display_name":"HyoBin","email":"hb.kim@okestro.com","username":"hbkim"},"ref":"refs/changes/70/903570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/70/903570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/70/903570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/70/903570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/70/903570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/70/903570/1"}}},"commit":{"parents":[{"commit":"4d606d4742df5a81afb22d57ce7eb152ca126abd","subject":"Adding Turkish translation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d606d4742df5a81afb22d57ce7eb152ca126abd"}]}],"author":{"name":"HyoBin Kim","email":"hb.kim@okestro.com","date":"2023-12-13 02:07:21.000000000","tz":540},"committer":{"name":"HyoBin Kim","email":"hb.kim@okestro.com","date":"2023-12-13 02:07:21.000000000","tz":540},"subject":"korean localization translation HMS DMY format","message":"korean localization translation HMS DMY format\n\nChange-Id: Ib985256ec8d8e6d12dcee147b2324271b233ad34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6480faa148270571bdca3448c17455376d74f082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6480faa148270571bdca3448c17455376d74f082"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
