)]}'
{"id":"openstack%2Fskyline-console~904538","triplet_id":"openstack%2Fskyline-console~master~Ic2f25fcc4620b06710c3ff1f26f84aba43881762","project":"openstack/skyline-console","branch":"master","topic":"styles","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-01-03 07:08:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic2f25fcc4620b06710c3ff1f26f84aba43881762","subject":"feat: support a custom UI theme","status":"MERGED","created":"2024-01-03 02:29:26.000000000","updated":"2024-01-03 07:09:35.000000000","submitted":"2024-01-03 07:08:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904538-styles","meta_rev_id":"460859edfe838d0fb9a07db41e82481ea25006d6","_number":904538,"virtual_id_number":904538,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-03 07:08:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-01-03 06:08:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-01-03 06:08:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-03 04:48:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-03 06:08:08.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"c24b440cf4901a5d696c7b94fc2b7cd26bd1d6f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-01-03 02:29:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"330f56ef30dd87766c38c8843e3fef1eda4b3665","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-01-03 03:28:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"162f4c79b10539f1f89b42a9badf13b40e07583f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-01-03 03:45:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d5c6d29e90674b85e4906a0aca67e03063f04136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 04:48:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/93adce950e794a889415261423dba36d\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/749b181343434092aeee8d2eef7b68e2 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7b23bef0c2f45beb6ec520c1ee3ec01 : SUCCESS in 4m 22s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3433c377db67491abce65ed8fca71578 : SUCCESS in 8m 46s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4f78508d05c841b1b9f2269c6cdb7ef7 : SUCCESS in 7m 15s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0e42963f159742719d4bfa2b9109b11b : SUCCESS in 7m 57s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1d1cb069d2054203b85f65df3552272a : SUCCESS in 1h 00m 43s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1181390b3d064a538d77c15589ced9c1 : SUCCESS in 49m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1025d10b1f7e42cc9cdb31779bca70b5 : FAILURE in 1h 01m 05s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/eec1ed6fc3ed4ac685e72b6b8663ad13 : SUCCESS in 42m 19s","accounts_in_message":[],"_revision_number":3},{"id":"f217294b5cbb5129d9e463026e3df83674e4e17d","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-01-03 04:55:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"42dc815dcd40f2265879e550b41456b8176278ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 06:06:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/194d5bb7f4c24e7db853cc4eb3cb003d\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b293406446d45b6954342b0f2a7963b : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e27abe2a6cdd43888f31f4851c6a93bf : SUCCESS in 5m 48s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d35f7e1523794f3985c3e1ff5c39bac7 : SUCCESS in 8m 28s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/372efc1a2e654bbba740c45d933f883f : SUCCESS in 5m 52s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/d582b41d5c9549febdff962816f5e240 : SUCCESS in 5m 03s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2566b2f589b2498ea4e807b0affcba4d : SUCCESS in 1h 02m 15s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2d738df6d558415bb04f31d7e831c2fa : SUCCESS in 50m 51s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a83f0d50cf2549b094f4d1cf67555e83 : SUCCESS in 45m 12s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2e22e96dc48347bcbd81cc51b5529f06 : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":3},{"id":"0ff6afa9102f396eff29cddc43e018100372d98a","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-01-03 06:08:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cafae2421b67a48fec0633a53eb9558f129ee7c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 06:08:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2fbaf9a1421bd67737bd14efc379d5f0421015fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 07:08:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d10268c2b50d4ccaaeea8a8160e9419c\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4e5b2718308485bb380f6848b1aafd4 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0913ae882b46429fc3302d5ffa9cab : SUCCESS in 4m 04s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/eaf32392b99e49d1a7c9e3cde96b59e2 : SUCCESS in 10m 58s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/dd31e76f7ef04f729dac522b0bf40c13 : SUCCESS in 4m 36s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/ac2815ac05fc4a89a8a6577c84404096 : SUCCESS in 4m 35s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0a3df523a67a41bdb9d1182f2b18c3c9 : SUCCESS in 59m 07s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7d97afb486aa4984bc59f9dec50e9ee5 : SUCCESS in 41m 16s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c44e437c2f07488f8634f93bca4c850e : SUCCESS in 40m 36s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d965018ed2a742e091de7857d1fa0ebb : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":3},{"id":"ea809f6341b5f1cca8df982d90fad79821d05d3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 07:08:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"460859edfe838d0fb9a07db41e82481ea25006d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 07:09:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6915a221fe934bd0a51b90f1e1a22d89\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20a6c5e126564f07ae28c95f55822f95 : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8137fe636d944a93b2e108c96fa8ad7c : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e0186599556a52012615b150e1e0d91f9c681aba","revisions":{"df746b402107d1d5c1c898804ff2971076d2bcd5":{"kind":"REWORK","_number":1,"created":"2024-01-03 02:29:26.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/38/904538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/38/904538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/904538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/904538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/904538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/38/904538/1"}}},"commit":{"parents":[{"commit":"f7041ca5a5dc8172c345a3fd4985d4c3e6470470","subject":"fix: Fix network editing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f7041ca5a5dc8172c345a3fd4985d4c3e6470470"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-16 01:32:30.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-01-03 02:29:09.000000000","tz":480},"subject":"feat: support a custom UI theme","message":"feat: support a custom UI theme\n\nsupport a custom UI theme\n\nChange-Id: Ic2f25fcc4620b06710c3ff1f26f84aba43881762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/df746b402107d1d5c1c898804ff2971076d2bcd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/df746b402107d1d5c1c898804ff2971076d2bcd5"}]},"branch":"refs/heads/master"},"150929a751740434e549f240cb11cfe0a1c2ced4":{"kind":"REWORK","_number":2,"created":"2024-01-03 03:28:57.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/38/904538/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/38/904538/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/904538/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/904538/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/904538/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/38/904538/2"}}},"commit":{"parents":[{"commit":"f7041ca5a5dc8172c345a3fd4985d4c3e6470470","subject":"fix: Fix network editing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f7041ca5a5dc8172c345a3fd4985d4c3e6470470"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-16 01:32:30.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-01-03 03:28:42.000000000","tz":480},"subject":"feat: support a custom UI theme","message":"feat: support a custom UI theme\n\nsupport a custom UI theme\n\nChange-Id: Ic2f25fcc4620b06710c3ff1f26f84aba43881762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/150929a751740434e549f240cb11cfe0a1c2ced4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/150929a751740434e549f240cb11cfe0a1c2ced4"}]},"branch":"refs/heads/master"},"e0186599556a52012615b150e1e0d91f9c681aba":{"kind":"REWORK","_number":3,"created":"2024-01-03 03:45:51.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/38/904538/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/38/904538/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/904538/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/904538/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/904538/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/38/904538/3"}}},"commit":{"parents":[{"commit":"f7041ca5a5dc8172c345a3fd4985d4c3e6470470","subject":"fix: Fix network editing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f7041ca5a5dc8172c345a3fd4985d4c3e6470470"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2023-11-16 01:32:30.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-01-03 03:45:28.000000000","tz":480},"subject":"feat: support a custom UI theme","message":"feat: support a custom UI theme\n\nsupport a custom UI theme\n\nChange-Id: Ic2f25fcc4620b06710c3ff1f26f84aba43881762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e0186599556a52012615b150e1e0d91f9c681aba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e0186599556a52012615b150e1e0d91f9c681aba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
