)]}'
{"id":"openstack%2Fskyline-console~905104","triplet_id":"openstack%2Fskyline-console~master~Ia9e7821f4f541ec13747b7632a2d69c58ec10285","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"36671":{"account":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"last_update":"2024-02-01 09:30:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia9e7821f4f541ec13747b7632a2d69c58ec10285","subject":"Adding Russian translation","status":"MERGED","created":"2024-01-09 11:12:00.000000000","updated":"2024-02-01 09:31:25.000000000","submitted":"2024-02-01 09:30:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"905104","meta_rev_id":"11de993c827b51b97b8c9ba1aa5d490ccdd7f644","_number":905104,"virtual_id_number":905104,"owner":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-01 09:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2024-02-01 07:55:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2024-02-01 07:55:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-09 13:15:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-17 09:28:55.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"CC"},{"updated":"2024-02-01 07:55:01.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"2d807b46219c33b2bd87a9c4ebd1f4c68657dd44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"date":"2024-01-09 11:12:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"867ad05c4e93fd83727a234f9af37418d66ab3d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 13:15:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b16c7764f95540fb88d59eda8cd72ea8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2309b6f799c942cbbe4728fc11bdc78e : SUCCESS in 4m 31s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d1ffa85ffbd049f3a0b90f91fb7c0770 : SUCCESS in 10m 32s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/7ade522369ab48ee82afacc684a0e59d : SUCCESS in 7m 19s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/cd608a4e604543edb5af481e7be81223 : SUCCESS in 7m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/572949f0da244447a732258334f72423 : SUCCESS in 1h 40m 14s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bac3e18139f841db8d205b30fdb86c7c : SUCCESS in 1h 16m 05s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7472602f67904490929827232008b884 : SUCCESS in 58m 28s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b4137ffbdf7e48549522c9cdbb2c222c : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":1},{"id":"57e757aeb01ddd5232d96edf95ea95cbeb1172ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"date":"2024-01-09 13:25:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"abc7f67ce9ddef58f1d0c3deb2a2e66975f2922f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 15:03:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/971d96330e7441f1ae031f0922b2ee19\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/652c94eb412c4c2abb895422f7345b35 : SUCCESS in 4m 59s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/57f4cd8c6ec248e280d79157bd2496cd : SUCCESS in 10m 54s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/deae95df3e9f4987b20c450cf12abbb6 : SUCCESS in 7m 13s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/363854d027534e6aafdf189d6cff6ca3 : SUCCESS in 6m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/05c0a2e41f9a49a89b3b54729bc6df98 : SUCCESS in 1h 32m 01s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ae37227cf8e64b2c847782fad5e8e656 : SUCCESS in 1h 04m 46s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/468047f22cce4281830729cba8116d9f : SUCCESS in 48m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e910d5ad224e4f35bbdda97516a99415 : SUCCESS in 1h 05m 12s","accounts_in_message":[],"_revision_number":2},{"id":"440196c604020b075976aeb57f1c20753df1f02e","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-01-17 09:28:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca0d2b9c040a890678c2cefcff8a2dc82a90da9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"date":"2024-01-20 23:54:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"dce05cbadc5dfb22bf5e94b72ce944fc94fcd035","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"date":"2024-01-21 00:22:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"81aae82142f54a9183c8ab777d99bb4b17dcbb49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"date":"2024-01-21 01:18:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"effbc0c21504c9a0266caa18cc90ece722a0061f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-21 01:58:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dab065e015794cbcad909c84c210958d\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/924a0119bcbd49f2b4ff238db79e2c1b : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3d91e8ed724263a910d1056e6571fb : SUCCESS in 4m 27s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4f6b700a596240e5a4b9a750c50179c0 : SUCCESS in 11m 14s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/efbddba29b654a76acae00102e93ee67 : SUCCESS in 7m 50s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/6f1ed6a295ed4fde9079ee3c9234ce57 : SUCCESS in 7m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2f1dfe99c328447c9a1a65111462f6d9 : FAILURE in 34m 45s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/113bf9db5bf84794a5668473699ddf1c : FAILURE in 13m 05s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5d596c9bf79d49f7ab710025a13c13c5 : FAILURE in 12m 19s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f2e4294e45784d1492c4faedc67f813d : FAILURE in 11m 37s","accounts_in_message":[],"_revision_number":5},{"id":"f2382e0981b0da425dfa2429ce6c91e1511cda43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"date":"2024-01-21 15:03:54.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"7db6e5f3adad64eec3ad31cec7bb308eafc85821","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-21 15:42:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a15da27733a4d2e9546ed6e03c29ab3\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fc0c8c99e5f498e9ea213f6d307a153 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4aa7e7a7f1a430aa326ede8daa9a99b : SUCCESS in 3m 50s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9fb24a489fb94f6fb1ca4d96db999aa8 : SUCCESS in 8m 10s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e68a503306c24eab856ef985b93bed40 : SUCCESS in 5m 29s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2e4528ca52fa446c8e35044ab46c6461 : SUCCESS in 7m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f3447164328a4f9cbf77f454b35de4be : FAILURE in 33m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d03aeb33588944f6b05bdbff359cf526 : FAILURE in 13m 01s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/16dd1d1d85f144fcb8fe935a053fba13 : FAILURE in 15m 35s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/bd1f3e43d97e479aaa9bcb94b330e376 : FAILURE in 15m 16s","accounts_in_message":[],"_revision_number":6},{"id":"23e3e1dc39812c350eb320d6f86fe7de8db87681","author":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"date":"2024-01-22 09:08:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7009f582373fe11e1fa0b0e90aa991798cdb03c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 10:09:49.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c1efb791f424695be74138c6a24c7df\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee5f79451e4b4fa88ca6e2582252b229 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75bad980ca0041bd9d04fc5f4439b050 : SUCCESS in 5m 35s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/69ca6529cf7046eab624ee4d7570cfc1 : SUCCESS in 7m 31s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/26228cc43d7e4032b3e98fc2fe7612d8 : SUCCESS in 5m 56s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/813434f4c16f4cb7a28e4a348bd043b3 : SUCCESS in 6m 45s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1ea2822e3ca546dba1d62b29b2858c63 : FAILURE in 35m 34s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1233e9f3524142caa37e25e9e3edcfc2 : FAILURE in 13m 36s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0bd6ae4cd63545b9a90d985ffb2e0346 : FAILURE in 27m 08s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/80b55e7d0fa9466985177d5c53ad56a3 : FAILURE in 12m 07s","accounts_in_message":[],"_revision_number":6},{"id":"f1255fc54131b06654f7c11b54368e5fb8f074cf","author":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"date":"2024-01-23 15:44:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6fb57d8c96c1dbd76b0925fa034e8e16b68b5cc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:29:46.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/390f3906931c4d57965caa4fcfd23758\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2257757d37b46d5ba53f9244340f9bf : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1af096f7aef46c3863bf2c2a96e45db : SUCCESS in 4m 41s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/9d33f597c59d46588b255c24f345c825 : SUCCESS in 11m 11s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/35ea474552594738af73996a1b6b9c98 : SUCCESS in 8m 10s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/df9e9d04763f4c80b69cc8f7b1a1deae : SUCCESS in 7m 07s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5d56baa9be9842bdb0f46254387582d3 : FAILURE in 38m 51s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/697e7281fbba4b5086e641eddd9b85c0 : FAILURE in 12m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2df9e6244a7e45569a60849e59800e21 : FAILURE in 25m 01s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/727e1c6100c1487b9b488e7cc01c3702 : FAILURE in 28m 18s","accounts_in_message":[],"_revision_number":6},{"id":"5b1cafe0a27eeaf6bdcac1d741c092d555f28d78","author":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"date":"2024-01-24 09:08:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b44b17435c504a53219e7ad285e5829193a40729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 11:40:33.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd1b3ba41a6a40f495b2ea8cd05d01d2\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e23550074fe46af9bc61ff811a0e000 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c42afa77ff241e69b417244cb2587f4 : SUCCESS in 4m 41s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/1e075fe5418f465bb6cd73c635c63ea2 : SUCCESS in 8m 55s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/16f72c95d0ac478b9ba756f5708057e0 : SUCCESS in 7m 31s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0c5fdebf66ba48a4842b997a85cc9765 : SUCCESS in 6m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5fce4320658844a4872f1f30f8efcb9f : FAILURE in 2h 16m 03s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bf3b073513a541c3ba2c87ff44e63305 : FAILURE in 1h 45m 49s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9609c1aafebf4f67869e83dc905a8ba5 : FAILURE in 2h 29m 47s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9e2da0b9457649b0a89a4fe154702af4 : FAILURE in 1h 46m 34s","accounts_in_message":[],"_revision_number":6},{"id":"f0d79d05b098596ff068d2b7c73a6adacebf59fe","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-01-25 01:49:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2cb463f2d3fa39fccd3b235f6fc9141e6bc6d940","author":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"date":"2024-01-26 12:54:50.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6425ce852312e1f731fba39627a571519e78785d","author":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"date":"2024-01-27 13:24:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"da39ea31fef2cf321f99094ae165488d44b49742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 16:03:50.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9c342d367a2418187ec92260bb0adda\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e85bf38c980c4b85b9ec8629a5f92ba1 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbb04427fd3541a4affae4c5650bc1d5 : SUCCESS in 4m 09s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/168398d2bb1e4c7eb09b678dec4a8373 : SUCCESS in 6m 24s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b6323f0b0a8b4524a05bf8fd480091a5 : SUCCESS in 7m 28s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/be21d99dc9b040be803a298020182fc0 : SUCCESS in 7m 51s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5e46245b59aa4487b46d608700e63b87 : FAILURE in 2h 11m 29s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/06b0205254984a45a7617beeb038dbb7 : FAILURE in 1h 41m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/41b9a039d2cd4de896c6b6523cc71039 : FAILURE in 2h 33m 20s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4d413ebbc46a4c38807b4f41f9c931c8 : FAILURE in 1h 45m 10s","accounts_in_message":[],"_revision_number":6},{"id":"2ca43928f7f97969c7384c78a788cee504e2bc02","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-02-01 06:11:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8adcb758ee6709939ba98ab02d5d57c724e93f18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 07:46:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3b9c073cb77431ca12602502053adc0\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d6e74fc6bf64858994edc413d19a0e4 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0603adb4432f42a3821d1ee4ab1ee10b : SUCCESS in 4m 43s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0da5211b5a0946e8907ecb4e6e8188ba : SUCCESS in 9m 58s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/864e1954c76149709737906d96c353c8 : SUCCESS in 7m 47s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/188803b9b2ec48d19422f81d2a77f25e : SUCCESS in 8m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d6d3f17e19874b0a8c2e68f9c840ef14 : SUCCESS in 1h 27m 48s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/6ded3932d5914d80af9d898514626a1d : SUCCESS in 1h 00m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/36b9bfc800a44d8e82a5a561f9cb3d82 : SUCCESS in 1h 02m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/64a454da999c4e399d745ba0bdaf7e2d : SUCCESS in 55m 35s","accounts_in_message":[],"_revision_number":6},{"id":"2472664ec5a9a734c7cc1e86659e6d12d42c4a92","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-02-01 07:55:01.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"5d26a57b89ef6da3ebd0a12b6d475eb65bb4d454","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 07:55:40.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ee42e0b322ec7bf4000b9a080137e1859adc47f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 09:30:22.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c50cbd1e2334a6ebea9e61902210c9f\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71542c3b2efd4522905fd3122fd6ce52 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210e037b70514b40a3fccd0968375614 : SUCCESS in 5m 25s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/56450c0e1a5f4e9680b94b355097505d : SUCCESS in 12m 45s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/fc607eb1da584ec0a82333b71003ab56 : SUCCESS in 7m 47s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/be35ec3d91b94710983865a2dd7830e7 : SUCCESS in 4m 34s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/96f373fe95ee4a3695f05340af3945f0 : SUCCESS in 1h 29m 02s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/840181b9f9fa4ff1ae135384b8d55463 : SUCCESS in 56m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7d4c44c4b4d04cc6b6a9ab2a9e974063 : SUCCESS in 59m 54s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/74eaa7fdc91d498b845a82097fce988e : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":6},{"id":"2d2e335a44dc23d6414bf4d4080be3ca9981eeef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 09:30:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"11de993c827b51b97b8c9ba1aa5d490ccdd7f644","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 09:31:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/507c671894fa450da5910031cbe0fba6\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ab9df24d47340bcaff33ebb2eae7307 : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbf7c2f58baf43ae8c838eb7976a7501 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c4e1f81a3c5b4a8ecdba4ec2a83f3a036039310f","revisions":{"e3a2df154d3a91dc1832689c97b1af73c639d7b7":{"kind":"REWORK","_number":1,"created":"2024-01-09 11:12:00.000000000","uploader":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"ref":"refs/changes/04/905104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/905104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/1"}}},"commit":{"parents":[{"commit":"e0186599556a52012615b150e1e0d91f9c681aba","subject":"feat: support a custom UI theme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e0186599556a52012615b150e1e0d91f9c681aba"}]}],"author":{"name":"Igor Bogomolov","email":"bogomolov@ispras.ru","date":"2024-01-09 10:58:48.000000000","tz":180},"committer":{"name":"Igor Bogomolov","email":"bogomolov@ispras.ru","date":"2024-01-09 11:05:49.000000000","tz":180},"subject":"Adding Russian translation","message":"Adding Russian translation\n\nChange-Id: Ia9e7821f4f541ec13747b7632a2d69c58ec10285\nSigned-off-by: Igor Bogomolov \u003cbogomolov@ispras.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3a2df154d3a91dc1832689c97b1af73c639d7b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3a2df154d3a91dc1832689c97b1af73c639d7b7"}]},"branch":"refs/heads/master"},"7dd5414cdc3e0efc3aec869fac26a975f78088f1":{"kind":"REWORK","_number":2,"created":"2024-01-09 13:25:01.000000000","uploader":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"ref":"refs/changes/04/905104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/905104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/2"}}},"commit":{"parents":[{"commit":"e0186599556a52012615b150e1e0d91f9c681aba","subject":"feat: support a custom UI theme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e0186599556a52012615b150e1e0d91f9c681aba"}]}],"author":{"name":"Igor Bogomolov","email":"bogomolov@ispras.ru","date":"2024-01-09 10:58:48.000000000","tz":180},"committer":{"name":"Igor Bogomolov","email":"bogomolov@ispras.ru","date":"2024-01-09 13:24:31.000000000","tz":180},"subject":"Adding Russian translation","message":"Adding Russian translation\n\nChange-Id: Ia9e7821f4f541ec13747b7632a2d69c58ec10285\nSigned-off-by: Igor Bogomolov \u003cbogomolov@ispras.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7dd5414cdc3e0efc3aec869fac26a975f78088f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7dd5414cdc3e0efc3aec869fac26a975f78088f1"}]},"branch":"refs/heads/master"},"489c5484b7b7497e407b0b744622ecd4860c683d":{"kind":"REWORK","_number":3,"created":"2024-01-20 23:54:38.000000000","uploader":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"ref":"refs/changes/04/905104/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/905104/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/3"}}},"commit":{"parents":[{"commit":"1919a36870f0e88344d9961c9b2f44f5ee191c77","subject":"Fix: fix error message of batch deleting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1919a36870f0e88344d9961c9b2f44f5ee191c77"}]}],"author":{"name":"Igor Bogomolov","email":"bogomolov@ispras.ru","date":"2024-01-20 23:42:30.000000000","tz":180},"committer":{"name":"Igor Bogomolov","email":"bogomolov@ispras.ru","date":"2024-01-20 23:53:17.000000000","tz":180},"subject":"Adding Russian translation","message":"Adding Russian translation\n\nChange-Id: Ia9e7821f4f541ec13747b7632a2d69c58ec10285\nSigned-off-by: Igor Bogomolov \u003cbogomolov@ispras.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/489c5484b7b7497e407b0b744622ecd4860c683d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/489c5484b7b7497e407b0b744622ecd4860c683d"}]},"branch":"refs/heads/master"},"c70906c6d9c10d7c7b17ad0c1ce0c75e752d3b8d":{"kind":"REWORK","_number":4,"created":"2024-01-21 00:22:17.000000000","uploader":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"ref":"refs/changes/04/905104/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/905104/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/4"}}},"commit":{"parents":[{"commit":"1919a36870f0e88344d9961c9b2f44f5ee191c77","subject":"Fix: fix error message of batch deleting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1919a36870f0e88344d9961c9b2f44f5ee191c77"}]}],"author":{"name":"Igor Bogomolov","email":"bogomolov@ispras.ru","date":"2024-01-20 23:42:30.000000000","tz":180},"committer":{"name":"Igor Bogomolov","email":"bogomolov@ispras.ru","date":"2024-01-21 00:22:03.000000000","tz":180},"subject":"Adding Russian translation","message":"Adding Russian translation\n\nChange-Id: Ia9e7821f4f541ec13747b7632a2d69c58ec10285\nSigned-off-by: Igor Bogomolov \u003cbogomolov@ispras.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c70906c6d9c10d7c7b17ad0c1ce0c75e752d3b8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c70906c6d9c10d7c7b17ad0c1ce0c75e752d3b8d"}]},"branch":"refs/heads/master"},"b8528fca77fd6ab027c07f758010d6c8abadcfe2":{"kind":"REWORK","_number":5,"created":"2024-01-21 01:18:01.000000000","uploader":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"ref":"refs/changes/04/905104/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/905104/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/5"}}},"commit":{"parents":[{"commit":"1919a36870f0e88344d9961c9b2f44f5ee191c77","subject":"Fix: fix error message of batch deleting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1919a36870f0e88344d9961c9b2f44f5ee191c77"}]}],"author":{"name":"Igor Bogomolov","email":"bogomolov@ispras.ru","date":"2024-01-20 23:42:30.000000000","tz":180},"committer":{"name":"Igor Bogomolov","email":"bogomolov@ispras.ru","date":"2024-01-21 01:16:52.000000000","tz":180},"subject":"Adding Russian translation","message":"Adding Russian translation\n\nChange-Id: Ia9e7821f4f541ec13747b7632a2d69c58ec10285\nSigned-off-by: Igor Bogomolov \u003cbogomolov@ispras.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b8528fca77fd6ab027c07f758010d6c8abadcfe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b8528fca77fd6ab027c07f758010d6c8abadcfe2"}]},"branch":"refs/heads/master"},"c4e1f81a3c5b4a8ecdba4ec2a83f3a036039310f":{"kind":"REWORK","_number":6,"created":"2024-01-21 15:03:54.000000000","uploader":{"_account_id":36671,"name":"Bogomolov Igor","display_name":"Bogomolov Igor","email":"bogomolov@ispras.ru","username":"bogomolov-igor"},"ref":"refs/changes/04/905104/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/905104/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/905104/6"}}},"commit":{"parents":[{"commit":"1919a36870f0e88344d9961c9b2f44f5ee191c77","subject":"Fix: fix error message of batch deleting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1919a36870f0e88344d9961c9b2f44f5ee191c77"}]}],"author":{"name":"Igor Bogomolov","email":"bogomolov@ispras.ru","date":"2024-01-20 23:42:30.000000000","tz":180},"committer":{"name":"Igor Bogomolov","email":"bogomolov@ispras.ru","date":"2024-01-21 15:03:18.000000000","tz":180},"subject":"Adding Russian translation","message":"Adding Russian translation\n\nChange-Id: Ia9e7821f4f541ec13747b7632a2d69c58ec10285\nSigned-off-by: Igor Bogomolov \u003cbogomolov@ispras.ru\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c4e1f81a3c5b4a8ecdba4ec2a83f3a036039310f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c4e1f81a3c5b4a8ecdba4ec2a83f3a036039310f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
