)]}'
{"id":"openstack%2Fskyline-console~906028","triplet_id":"openstack%2Fskyline-console~master~I4858d9315397fa2833f34e0ec56d634780946755","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2024-01-18 10:19:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4858d9315397fa2833f34e0ec56d634780946755","subject":"Fix: fix error message of batch deleting","status":"MERGED","created":"2024-01-18 08:04:25.000000000","updated":"2024-01-18 10:20:34.000000000","submitted":"2024-01-18 10:19:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906028","meta_rev_id":"b4cacbe6d7d7ae58aac7dcfcd9a07a5603e8265f","_number":906028,"virtual_id_number":906028,"owner":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 10:19:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2024-01-18 09:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2024-01-18 09:08:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-18 09:01:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-18 09:07:48.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"d5790290bfdcd68cc37343c011d16104df6d1aab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2024-01-18 08:04:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a359185fcd9b5e18e9b0d201d996f1d4480a034","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 09:01:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d75e2c2a2d24a5d8f30330699c2120b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47909b7d573e412e807fbd34aad0b285 : SUCCESS in 5m 20s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/6dd665d77d0640f193d098748de41d5d : SUCCESS in 10m 28s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6eb0415695334c29af5623a5319c82dc : SUCCESS in 7m 11s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/ad28171eb2624a9e94f751e33cd1eac8 : SUCCESS in 7m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/73cbe708b787444e98263b880e512d8d : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a1eb4c332ae7c3b5480c5253d98bdb42ac34a1b2","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-01-18 09:07:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df6e29d43126fe3fec35b6242e1c4cb263cecb16","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-01-18 09:08:01.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b64e466ee8ed06f0425bffd9a3f7c27fab7dad6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 09:08:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a8975cabda7c1ea51050a9c1c8187289637d7ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 10:19:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0addd167b5774d5c8ceccdbd25cea9fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c543f6ecb78249a0bf93d1620ef1e432 : SUCCESS in 4m 05s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8e57b9b5206c460f8cce326ffc3b0519 : SUCCESS in 11m 14s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ad028e022d5148aeac62e68db11fed77 : SUCCESS in 7m 36s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/544c6205d65e45b492709598abe08bbc : SUCCESS in 7m 30s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2e91c047854f4a7ba70715988cdf1481 : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1a3b50b866bde2f35e3e2b567288e51cd9f3cb29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 10:19:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b4cacbe6d7d7ae58aac7dcfcd9a07a5603e8265f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 10:20:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01d2596427a34aefb422e4fdda81c4f3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35495a34d4ba4171ab6476ffc1cbce79 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1919a36870f0e88344d9961c9b2f44f5ee191c77","revisions":{"1919a36870f0e88344d9961c9b2f44f5ee191c77":{"kind":"REWORK","_number":1,"created":"2024-01-18 08:04:25.000000000","uploader":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"ref":"refs/changes/28/906028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/28/906028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/906028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/906028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/28/906028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/28/906028/1"}}},"commit":{"parents":[{"commit":"0dd9dd0062c0faf1c66d23f87d74ede4a8ffd006","subject":"Merge \"korean localization translation HMS DMY format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0dd9dd0062c0faf1c66d23f87d74ede4a8ffd006"}]}],"author":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2024-01-18 08:04:04.000000000","tz":480},"committer":{"name":"xusongfu","email":"xu.songfu@99cloud.net","date":"2024-01-18 08:04:04.000000000","tz":480},"subject":"Fix: fix error message of batch deleting","message":"Fix: fix error message of batch deleting\n\nFix volume name for volume snapshot batch deleting\n\nChange-Id: I4858d9315397fa2833f34e0ec56d634780946755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1919a36870f0e88344d9961c9b2f44f5ee191c77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1919a36870f0e88344d9961c9b2f44f5ee191c77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
