)]}'
{"id":"openstack%2Fskyline-console~907260","triplet_id":"openstack%2Fskyline-console~master~Ia780bca1476c98d699059ee60755c0636997067d","project":"openstack/skyline-console","branch":"master","topic":"simple-table","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-02-01 10:11:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia780bca1476c98d699059ee60755c0636997067d","subject":"feat: support copyable for SimpleTable component","status":"MERGED","created":"2024-01-31 02:27:54.000000000","updated":"2024-02-01 10:12:49.000000000","submitted":"2024-02-01 10:11:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907260-simple-table","meta_rev_id":"8aa66e35c1d2cc1e5ad9b6558471350759860b75","_number":907260,"virtual_id_number":907260,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-01 10:11:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-02-01 08:48:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-02-01 08:48:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-31 05:53:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-01 08:48:59.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"a08c613562adffaa665905c5b1c0d718387f8e27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-01-31 02:27:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"843189f8bd1d027d1ac297f8a56b1351c8a66fd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 05:53:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05041f9743614722875e0b3241d2c348\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d52b5a8090d0460a959c28201f6d5bf0 : SUCCESS in 6m 46s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8c3bad7f13cc4fc89bc7dcb24949c61c : SUCCESS in 7m 44s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/da691337324a4b3999b99a963a473bde : SUCCESS in 7m 14s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/870822c422fc4499b0e3533ba1b05b49 : SUCCESS in 8m 13s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/406e3419755b4b80a51638c80bb78bdc : FAILURE in 2h 21m 03s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a5e1a6fb3c66453782b2ea5d1f8259ae : FAILURE in 1h 28m 53s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/555377475fe247fc857d97fca3a0e5f2 : FAILURE in 2h 19m 58s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d9fe9bcfe82f4ed8b0b1a085b231e391 : FAILURE in 1h 33m 26s","accounts_in_message":[],"_revision_number":1},{"id":"86d04ba43e08d6baaecee851240478e6b1720ace","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-02-01 06:57:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d20b39cd29847415d5e9c5f28affdee67b535595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 08:33:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0316b2da6eb2452b9d33eee37070e2ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd12618836614ad095dcac213cfc5d4e : SUCCESS in 7m 26s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/55bc1a31ec474679855dc0e97714391e : SUCCESS in 13m 19s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0131f71f87f14330b0e3478f51fbdf48 : SUCCESS in 8m 04s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/42b68a2649464d07b28ea93299fffe5e : SUCCESS in 4m 57s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3357d20a86874224a8c7d8aa16a2f12d : SUCCESS in 1h 29m 18s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/cd895e8b1c0f4eec945f3a0579da7439 : SUCCESS in 58m 07s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/173b4afb8bf346e1b3cf92d32cb6d041 : SUCCESS in 1h 02m 26s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a01a917a24ba418aa1ca7f671efaf6ac : SUCCESS in 38m 51s","accounts_in_message":[],"_revision_number":1},{"id":"28c9be1058c2ead50c7dffafeb712bb804cfac4d","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-02-01 08:48:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1576e806a969a496ab1e0115b5e52ad7b4081f3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 08:49:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f57c8ea3202732b103a8d5d188535508e1c25d2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 10:11:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc9417939eea40049b2dde1d83d760b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bf455f63934457ba686385bc03df82c : SUCCESS in 4m 42s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/bed43716897740e4bda3e272ba176e10 : SUCCESS in 9m 33s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/d55a0c2a0e724a59b7750c2e25e4e3a7 : SUCCESS in 6m 39s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/350f08358e8e44f199895dc7baa2c5e6 : SUCCESS in 7m 01s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2ca70d7e320943e09e72fd1b2f49b9a1 : SUCCESS in 1h 17m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/cf751345d98145c4968342aaaad20496 : SUCCESS in 58m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bdf8c4fd963845aaba14a3a5b848168b : SUCCESS in 56m 42s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3ea9ca08f40e4230934c873bc714e6b3 : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0ddc19a85a8526fbe0f0d2bdd13415138605cb74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 10:11:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8aa66e35c1d2cc1e5ad9b6558471350759860b75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 10:12:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e62bd1a4b6c4842a3b3c019140196c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be261dec0d764c3490f880dc53d4115a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c97ddb3cee8d53dc9614509bf1f3f7197cddcea","revisions":{"7c97ddb3cee8d53dc9614509bf1f3f7197cddcea":{"kind":"REWORK","_number":1,"created":"2024-01-31 02:27:54.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/60/907260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/60/907260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/60/907260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/60/907260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/60/907260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/60/907260/1"}}},"commit":{"parents":[{"commit":"8a1d7ab034c174a2b216109438c12e3ae70c1af3","subject":"refactor: update icon render for menu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8a1d7ab034c174a2b216109438c12e3ae70c1af3"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-01-31 02:27:36.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-01-31 02:27:36.000000000","tz":480},"subject":"feat: support copyable for SimpleTable component","message":"feat: support copyable for SimpleTable component\n\nSupport copyable attribute for SimpleTable component\n\nChange-Id: Ia780bca1476c98d699059ee60755c0636997067d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7c97ddb3cee8d53dc9614509bf1f3f7197cddcea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7c97ddb3cee8d53dc9614509bf1f3f7197cddcea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
