)]}'
{"id":"openstack%2Fskyline-console~907287","triplet_id":"openstack%2Fskyline-console~master~I66bdbf848d375e45f0ab8941b0989165ad86a137","project":"openstack/skyline-console","branch":"master","topic":"select-table","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-02-01 10:31:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66bdbf848d375e45f0ab8941b0989165ad86a137","subject":"feat: support refresh button for SelectTable component","status":"MERGED","created":"2024-01-31 06:23:42.000000000","updated":"2024-02-01 10:33:02.000000000","submitted":"2024-02-01 10:31:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907287-select-table","meta_rev_id":"09a238d8307bf2529d6e6fbc50318b582d666ae2","_number":907287,"virtual_id_number":907287,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-01 10:31:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-02-01 08:49:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":1,"date":"2024-01-31 06:26:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-02-01 08:49:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-31 06:26:49.000000000","updated_by":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2024-01-31 09:00:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-01 08:49:15.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"575c5fa3d555c3e242f95b9c43a6345e0775ef79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-01-31 06:23:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3441d6fdb180cf10c24448064127cd13124fd69","author":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"date":"2024-01-31 06:26:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e43ed1d5dcc5c1bbb8cd2f494a803657ad74ca5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 09:00:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cb2642334e7424baa577d0c32acc6f5\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba98313ad0ef41deb8a651756aacabd4 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/497aea75b1274908a1262ca538fd888f : SUCCESS in 3m 44s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/75cbf36c8a0e43f29a56f521ad4e8f9c : SUCCESS in 11m 05s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/56fc19da0bdd44bb9ec18536f82ba941 : SUCCESS in 4m 33s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/63e3d4b0e709467e9f2de07194480a89 : SUCCESS in 9m 03s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4c854c63918f48cb83859915bf07acc1 : FAILURE in 1h 55m 25s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7f237ea9f627487ebc82ad0f5e2d9cf9 : FAILURE in 1h 42m 20s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/60d79babed6743f7a1e7882726028969 : FAILURE in 2h 35m 59s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/99192c72c2d54f018ae60a4d162d11ff : FAILURE in 1h 52m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c315265bcc57f92708ab9bdb209bdc4b5b126501","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-02-01 06:58:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2254b297d10382ea26945dc08852e21dc8c09f18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 08:15:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/044308755a714224a91ffb8a32e2b59a\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41611c522918430a925fac9d41150713 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc984c4efa2c4dabb32228b31a291b41 : SUCCESS in 5m 16s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c20e89b07d2941bea35a8a3fe8fd80f4 : SUCCESS in 10m 17s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/f0a2c19f2a9f4727b11ddb83c917fea1 : SUCCESS in 6m 27s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/499fe8fd154446cb9d16fbcd3baef9c4 : SUCCESS in 6m 28s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f888acb67cf04970bf359bf356112f24 : SUCCESS in 1h 12m 50s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1820824cc5ec4cac80fdff660f8c23ba : SUCCESS in 54m 00s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/17a7bc4db4c8494bb2c8a96bce310143 : SUCCESS in 56m 06s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1687c8c7a73e49558c79b299edbcfcbb : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":1},{"id":"977ef1dccf904942ffcd731e4061433246aca0cd","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-02-01 08:49:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d48e9eb2f1a6464f5c8458570bc2cfb3e2be443","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 08:49:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a73267ff2b5077d257a99d25499a27045af1d289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 10:31:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc74e1504e6f4646bfdbece6c31f185c\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b069c6f5b0074f28b513491a9a458d6e : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c869774fa814cc5ad70acc7a39cee31 : SUCCESS in 4m 27s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f5fc6fbf5937437aa9eb1b23cd11de57 : SUCCESS in 11m 33s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/8ba59d8b879244d892f0d7e833d1a79c : SUCCESS in 8m 26s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/3f6f492718684a8eb7cb5467f4ffac6b : SUCCESS in 6m 00s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7dd5bbecb24c4fd1aeb428b4dc5455c0 : SUCCESS in 1h 35m 36s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9ff81e85195c4000ab8c785a77e2f410 : SUCCESS in 45m 59s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5a8584af3d4a4fa7af5d4522f21c3430 : SUCCESS in 52m 51s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c4b510559042417b9cc53d2b75abdb62 : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f794da92fa2428c463b865cf8c13da7c21a15580","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 10:31:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"09a238d8307bf2529d6e6fbc50318b582d666ae2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 10:33:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f33382835ca4a7185c5d57b80347f68\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0038c47278849e496e85638ef417335 : SUCCESS in 59s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a1d0ae8d14647df90a9b4f92a73a84e : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5fcc3b053ea8bbc42eb73abfff3c4c106f07feb6","revisions":{"5fcc3b053ea8bbc42eb73abfff3c4c106f07feb6":{"kind":"REWORK","_number":1,"created":"2024-01-31 06:23:42.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/87/907287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/87/907287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/907287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/907287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/87/907287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/87/907287/1"}}},"commit":{"parents":[{"commit":"8a1d7ab034c174a2b216109438c12e3ae70c1af3","subject":"refactor: update icon render for menu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8a1d7ab034c174a2b216109438c12e3ae70c1af3"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-01-31 06:21:55.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-01-31 06:21:55.000000000","tz":480},"subject":"feat: support refresh button for SelectTable component","message":"feat: support refresh button for SelectTable component\n\n1. Support refresh button to refresh data for the SelectTable component.\n2. When creating an instance/ironic/container, you can click to create a network/security group, then create a new resource in a new window, refresh the data in the previous window and select the newly created resource.\n\nCloses-Bug: 2042928\nChange-Id: I66bdbf848d375e45f0ab8941b0989165ad86a137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5fcc3b053ea8bbc42eb73abfff3c4c106f07feb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5fcc3b053ea8bbc42eb73abfff3c4c106f07feb6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
