)]}'
{"id":"openstack%2Fskyline-console~907512","triplet_id":"openstack%2Fskyline-console~master~I92a9101e63114dddee0810089a2596c124282102","project":"openstack/skyline-console","branch":"master","topic":"select-table","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-02-02 08:24:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I92a9101e63114dddee0810089a2596c124282102","subject":"feat: open new window for creating network/subnet","status":"MERGED","created":"2024-02-02 02:28:55.000000000","updated":"2024-02-02 08:25:38.000000000","submitted":"2024-02-02 08:24:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907512-select-table","meta_rev_id":"d2e262d956245e5577414173a73cf6d0d4cb493c","_number":907512,"virtual_id_number":907512,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-02 08:24:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-02-02 06:32:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-02-02 06:32:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-02 06:27:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-02 06:32:57.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"6eb725ec84ccbab0d706e5624b9c291a37fc1f41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-02-02 02:28:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e07792a94a08a6360a3afe45fbc66f18955ee680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 06:27:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b165a7815274d18bf4ebbc7067cd42c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/855eb2fdbdcc4605a432b609daf16ea0 : SUCCESS in 5m 01s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/f8b8c6d9877f454b95f16118daebaa91 : SUCCESS in 10m 14s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/4fc24f910e9f436d986af4ce6d389710 : SUCCESS in 7m 38s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/536793d14142479fbc63ee51729cefe3 : SUCCESS in 8m 12s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/41556965a7d34a95bf0a13b3d187d304 : SUCCESS in 1h 27m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c3ba830fa2418de82355873c92436a46dc0d464d","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-02-02 06:32:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4919ce63ccaf38a3011663f0bab8033b0e0dbb0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 06:33:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6be2ddcca356db953a82f14a3b4cf2f2359df564","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 08:24:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18899b15543f4e39a0451f506b2c443c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac8751d0d4549ba9b862061a0a23667 : SUCCESS in 5m 19s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3ab80bba83884737bacaab21cdd80384 : SUCCESS in 11m 28s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/96d4aaa8bcab4fb29df5183cfdb9da99 : SUCCESS in 7m 48s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/faafecd8c57741898991ea2919bc014a : SUCCESS in 5m 35s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5acbc18e918245ee8e2e610c75fc94d9 : SUCCESS in 1h 38m 12s","accounts_in_message":[],"_revision_number":1},{"id":"638d51813b3f170d730b62fa12a72fcde000d914","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 08:24:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d2e262d956245e5577414173a73cf6d0d4cb493c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 08:25:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ff73f7f84e14487845fe2fd4c1dee4d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4558931b10e94fd284fd0bf18937a52e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8793a195d30526a2893f0d7dc736a74d8ff1eb3","revisions":{"b8793a195d30526a2893f0d7dc736a74d8ff1eb3":{"kind":"REWORK","_number":1,"created":"2024-02-02 02:28:55.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/12/907512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/12/907512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/12/907512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/12/907512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/12/907512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/12/907512/1"}}},"commit":{"parents":[{"commit":"5fcc3b053ea8bbc42eb73abfff3c4c106f07feb6","subject":"feat: support refresh button for SelectTable component","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5fcc3b053ea8bbc42eb73abfff3c4c106f07feb6"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-02-02 02:28:21.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-02-02 02:28:21.000000000","tz":480},"subject":"feat: open new window for creating network/subnet","message":"feat: open new window for creating network/subnet\n\nOpen a new window for createing network/subnet when in the ironic creation page.\n\nChange-Id: I92a9101e63114dddee0810089a2596c124282102\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b8793a195d30526a2893f0d7dc736a74d8ff1eb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b8793a195d30526a2893f0d7dc736a74d8ff1eb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
