)]}'
{"id":"openstack%2Fskyline-console~910871","triplet_id":"openstack%2Fskyline-console~master~Ibee94fd5895c0dfd295a3b735aac46f428cbb5b9","project":"openstack/skyline-console","branch":"master","topic":"bucket","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-03-04 03:56:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibee94fd5895c0dfd295a3b735aac46f428cbb5b9","subject":"fix: fix catalog region display","status":"MERGED","created":"2024-03-04 02:09:10.000000000","updated":"2024-03-04 03:57:57.000000000","submitted":"2024-03-04 03:56:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910871-bucket","meta_rev_id":"5b323e02c2e8e4d9671453fcbaaef05bd192ea09","_number":910871,"virtual_id_number":910871,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-04 03:56:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-03-04 03:12:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-03-04 03:12:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-04 03:07:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-04 03:12:02.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"bfac3d4bbab86e038d2fa6af649277991c9ca264","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-03-04 02:09:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe17cdff96e7e24147442568c5b00cd28db16658","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 03:07:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a11a8327e66249fda225320cf63396c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c88993fa05614ff48dacc2f2823b62c5 : SUCCESS in 6m 14s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/1ceb8c07b27d4426b808a8bb3fa1b04f : SUCCESS in 11m 04s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/bba261413a7d4133bcec346f2a93ad52 : SUCCESS in 6m 08s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f5033d70de6a4a9aa3b3e84c489e9c5c : SUCCESS in 8m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d02c14932f3d45e79a90cccc86ae3b0f : SUCCESS in 50m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c0d2baa188729a8e798b2e8a4e0c0bff8b57da0c","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-03-04 03:12:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f57e3875c7f38ea52de5a5cc2fc9aa6ca39790a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 03:12:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d76b1297df1fd757a0a653f818158698eb6aba4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 03:56:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0df213dec94494f82f0e5319c01617a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1130750c4fde412cbb2671b0a6f779d8 : SUCCESS in 4m 55s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/468cc916f4e74f76ab703248a8b18f5e : SUCCESS in 11m 03s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/2ca8d89f71d74ab7a75e332c953e1472 : SUCCESS in 4m 58s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2957aa98d5d84cef916e8ff4dfa7cd62 : SUCCESS in 5m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e6c6ebafd9864bc8a20839b58c1ca18b : SUCCESS in 43m 36s","accounts_in_message":[],"_revision_number":1},{"id":"41c128269b0cc21583ae085469753749bb46c090","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 03:56:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5b323e02c2e8e4d9671453fcbaaef05bd192ea09","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 03:57:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff177650d743413e93282108b3828039\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8216c5f5a0a420bb4ab1a6bcf78d060 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b42958e43db76d75dc713d27917856b5a4a48228","revisions":{"b42958e43db76d75dc713d27917856b5a4a48228":{"kind":"REWORK","_number":1,"created":"2024-03-04 02:09:10.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/71/910871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/71/910871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/71/910871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/71/910871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/71/910871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/71/910871/1"}}},"commit":{"parents":[{"commit":"8c20aa25480498ae2d7db2fb31d9adfecb1329be","subject":"feat: update scrollbar style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8c20aa25480498ae2d7db2fb31d9adfecb1329be"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-04 02:08:21.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-04 02:08:56.000000000","tz":480},"subject":"fix: fix catalog region display","message":"fix: fix catalog region display\n\nfix the region display when the catalog\u0027s endpoints is empty\n\nChange-Id: Ibee94fd5895c0dfd295a3b735aac46f428cbb5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b42958e43db76d75dc713d27917856b5a4a48228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b42958e43db76d75dc713d27917856b5a4a48228"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
