)]}'
{"id":"openstack%2Fskyline-console~910873","triplet_id":"openstack%2Fskyline-console~master~Iece38c32c681fd07e69d42f3f3f9901fee2bdf8f","project":"openstack/skyline-console","branch":"master","topic":"instance","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-03-04 09:56:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iece38c32c681fd07e69d42f3f3f9901fee2bdf8f","subject":"fix: fix set security group when create instance","status":"MERGED","created":"2024-03-04 06:45:21.000000000","updated":"2024-03-04 09:57:10.000000000","submitted":"2024-03-04 09:56:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910873-instance","meta_rev_id":"d4550ae1825ea9399bc47ff0a98dfe8add7c2620","_number":910873,"virtual_id_number":910873,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-04 09:56:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-03-04 08:25:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-03-04 08:25:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-04 08:12:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-04 08:25:57.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"9019fdd82888863a4e3df0daa2d3378c5e1ec2fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-03-04 06:45:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"609f0c84c591f20e3514810ed443e7edcc71d66f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 08:12:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43189cc6ee2b4b998e0bd98fd48458f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65332a51eaa7498d8ac7c5dbe70c4018 : SUCCESS in 3m 49s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/a124b4cb3fa34b7c99b1400d29ca869d : SUCCESS in 9m 06s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/9771231b04294677afe8b1a908dcbe5a : SUCCESS in 6m 50s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/5b8ab9f0a84146238e6cfaededc70d82 : SUCCESS in 5m 50s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/03ed92d45b0e4a3f9dfefd568c7eee97 : SUCCESS in 1h 21m 24s","accounts_in_message":[],"_revision_number":1},{"id":"18a0e455f8b98b0758c23d135a6d264ec24d0a8b","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-03-04 08:25:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fbcced5dbe982fc94db19fe40f7ff60bcc5f0873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 08:26:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93d7eeb02f5365d90b4dec651ef7dfda296b0b7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 09:56:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/135b7feed68f4981b6847ed0f2ffb33f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a60730006854204a5411f6afb72bcf6 : SUCCESS in 4m 10s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/48ad505964f6447b84a0d3c8108233ce : SUCCESS in 10m 14s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/360aabf4ae054caa8b00e4d3fde0689a : SUCCESS in 5m 29s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/e1b7afb1a2984339be6909660a8b7fc9 : SUCCESS in 9m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3a26396f6c474326afc40ab887523589 : SUCCESS in 1h 24m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e6c79e1199622ce675480b2c62b78441f8b306be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 09:56:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d4550ae1825ea9399bc47ff0a98dfe8add7c2620","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 09:57:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cd2b5aebe644d948c5ee3355b4f8fdb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b378f891179548daa27df30dbf88de84 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ade5757898aafc94e9df011aa40005e7dfc0c67f","revisions":{"ade5757898aafc94e9df011aa40005e7dfc0c67f":{"kind":"REWORK","_number":1,"created":"2024-03-04 06:45:21.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/73/910873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/73/910873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/73/910873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/73/910873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/73/910873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/73/910873/1"}}},"commit":{"parents":[{"commit":"b42958e43db76d75dc713d27917856b5a4a48228","subject":"fix: fix catalog region display","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b42958e43db76d75dc713d27917856b5a4a48228"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-04 06:43:14.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-04 06:44:10.000000000","tz":480},"subject":"fix: fix set security group when create instance","message":"fix: fix set security group when create instance\n\nHide security group selector when select any network or port which the port_security_enabled is false\n\nChange-Id: Iece38c32c681fd07e69d42f3f3f9901fee2bdf8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ade5757898aafc94e9df011aa40005e7dfc0c67f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ade5757898aafc94e9df011aa40005e7dfc0c67f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
