)]}'
{"id":"openstack%2Fskyline-console~911635","triplet_id":"openstack%2Fskyline-console~master~Ib2955fbb5b9d9cfb00eacd9a6ad9c45a83a0c7c5","project":"openstack/skyline-console","branch":"master","topic":"volume","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-03-07 05:56:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib2955fbb5b9d9cfb00eacd9a6ad9c45a83a0c7c5","subject":"fix: disable delete mounted volume","status":"MERGED","created":"2024-03-07 03:03:42.000000000","updated":"2024-03-07 05:57:12.000000000","submitted":"2024-03-07 05:56:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911635-volume","meta_rev_id":"0b64955ce4bec2f7369b07bf79cbd5001c9a5618","_number":911635,"virtual_id_number":911635,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-07 05:56:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-03-07 04:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-03-07 04:57:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-07 04:06:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-07 04:57:10.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"355d02f69f16717a836f646d2cfd463ae449a4bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-03-07 03:03:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92dfbbb46b58fa099b8fd0d6181c8aa5cae86e8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 04:06:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1478f631e88f4361b4e523b3c680bec8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/560cd5fe5813489a845713bb027fd2bc : SUCCESS in 4m 04s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4296fb89af864db3934e389fe1ea1b2f : SUCCESS in 11m 25s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/1ca2fa2137d54742895d9fd728366d91 : SUCCESS in 7m 23s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/f7a40ef85b564704ab5f1d5af40e31e9 : SUCCESS in 6m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f5c4f1bfd9f74ed78e684e751c08f0c2 : SUCCESS in 56m 34s","accounts_in_message":[],"_revision_number":1},{"id":"4339ed02fb45e8f83d4036adb3ffedb34fd6dc07","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-03-07 04:57:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ebf568c8347ae52caa50fbd5d5d4583992391d67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 04:57:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79540d5bd9c075c8ae7c4d3b78a378a1457b140a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 05:56:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fee3f90ae65435aafcb520b03cbd610\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c45d555d65a40878e16c121dce03dbc : SUCCESS in 4m 28s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d644688b03104193b225a48ae1fdc463 : SUCCESS in 10m 44s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/cf4089398a1642848060537ddb52dd91 : SUCCESS in 7m 01s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/0638e6a63b8e449ab2c65d86056b50ad : SUCCESS in 7m 26s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f8e8c8df78db47d1b376812a6d45437f : SUCCESS in 53m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ade29cc59e9329332dd37d0d977741dcdbe7cec6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 05:56:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0b64955ce4bec2f7369b07bf79cbd5001c9a5618","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 05:57:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0d06befd2e540b392d093f7ccab58ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c462921c4a3149b89b8898336a3a09ba : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffc75102adaecaffa6cd9d38fff91cc50f1ec740","revisions":{"ffc75102adaecaffa6cd9d38fff91cc50f1ec740":{"kind":"REWORK","_number":1,"created":"2024-03-07 03:03:42.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/35/911635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/35/911635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/911635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/911635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/35/911635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/35/911635/1"}}},"commit":{"parents":[{"commit":"48444e9ee7408f0a96f57d7d25557651406a7f57","subject":"feat: update i18n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/48444e9ee7408f0a96f57d7d25557651406a7f57"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-07 03:01:27.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-07 03:01:50.000000000","tz":480},"subject":"fix: disable delete mounted volume","message":"fix: disable delete mounted volume\n\nDisable delete volume when the volume is error and is mounted to some instance\n\nChange-Id: Ib2955fbb5b9d9cfb00eacd9a6ad9c45a83a0c7c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ffc75102adaecaffa6cd9d38fff91cc50f1ec740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ffc75102adaecaffa6cd9d38fff91cc50f1ec740"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
