)]}'
{"id":"openstack%2Fskyline-console~911644","triplet_id":"openstack%2Fskyline-console~master~Ia7db4977dc80ae23a2e40cef64310f290b48e5a6","project":"openstack/skyline-console","branch":"master","topic":"port","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-03-08 01:54:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia7db4977dc80ae23a2e40cef64310f290b48e5a6","subject":"feat: update the network column of the port list","status":"MERGED","created":"2024-03-07 06:57:35.000000000","updated":"2024-03-08 01:55:19.000000000","submitted":"2024-03-08 01:54:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911644-port","meta_rev_id":"0b03e7c4221565f1e7ca6998204a27a213b3e449","_number":911644,"virtual_id_number":911644,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-08 01:54:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-03-08 01:06:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-03-08 01:06:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-07 08:05:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-08 01:06:52.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"9337d231b92474817e97a1f7412c493f33762850","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-03-07 06:57:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71e585fa90b4bdb5aa802ea9ca83dc435145530c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 08:05:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd12000d4f3b4e549bd2694596669c85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f5744e20bbd449691453d21c89cbdce : SUCCESS in 4m 33s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/ff74adf62eb944299183d971be13bb54 : SUCCESS in 10m 26s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/62e14c4281fc471f8213167490d4c8e8 : SUCCESS in 6m 10s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/fb24b9a4d6ea403f9f6b117ff7c81929 : SUCCESS in 7m 15s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/01aeb823527c42a4899a6d1bbd36bdc9 : SUCCESS in 1h 02m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2aead5a781b25b25627ea69a15dd2bc0844b06ce","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-03-08 01:06:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64c907fefa1d5079f26ab241d24696a21f66aa35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 01:07:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93f5748f005e8a2969cd75b272740f0376052016","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 01:54:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9a921ff7d1d4128bf8b29edcee34a75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3f7036e2ab488a9db861ad5101f292 : SUCCESS in 4m 37s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0437a70bd41342b996db609cd8411d48 : SUCCESS in 6m 06s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/0dde1c20ec3840f18c534bdd183f8d6f : SUCCESS in 4m 22s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/50ff0659d9074c5695b489eef4e906c0 : SUCCESS in 6m 10s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/beebd738ca9444459491f845c7b38412 : SUCCESS in 46m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b4311c849932dd340d587178a61cd97f7fe81f31","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 01:54:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0b03e7c4221565f1e7ca6998204a27a213b3e449","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 01:55:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8db95935806f45f6a4c02e191ebe6ca3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a33d3128e0f4928bc945e7c0c467d79 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc99a8199f2b5694a6341d3df972e7b5faa9b996","revisions":{"cc99a8199f2b5694a6341d3df972e7b5faa9b996":{"kind":"REWORK","_number":1,"created":"2024-03-07 06:57:35.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/44/911644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/44/911644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/911644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/911644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/911644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/44/911644/1"}}},"commit":{"parents":[{"commit":"17221ece1feb4f2e67e75061b5546494c8a6fcb0","subject":"Merge \"feat: update external network column title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/17221ece1feb4f2e67e75061b5546494c8a6fcb0"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-07 06:55:37.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-07 06:56:29.000000000","tz":480},"subject":"feat: update the network column of the port list","message":"feat: update the network column of the port list\n\nUpdate the network column title of the port list, so that the download file contains the network id.\n\nChange-Id: Ia7db4977dc80ae23a2e40cef64310f290b48e5a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc99a8199f2b5694a6341d3df972e7b5faa9b996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc99a8199f2b5694a6341d3df972e7b5faa9b996"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
