)]}'
{"id":"openstack%2Fskyline-console~913738","triplet_id":"openstack%2Fskyline-console~master~I42005ff85d5c329aaf4a18a9d6181e901d091217","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"20865":{"account":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"last_update":"2024-03-26 08:50:14.000000000","reason":"\u003cGERRIT_ACCOUNT_20865\u003e replied on the change","reason_account":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2024-03-26 11:39:07.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-03-26 09:57:34.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2024-03-26 11:39:07.000000000","reason":"Change was submitted"},"34671":{"account":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"last_update":"2024-03-26 11:39:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I42005ff85d5c329aaf4a18a9d6181e901d091217","subject":"feat: Adding log feature to Nova instance detail page.","status":"MERGED","created":"2024-03-20 08:53:32.000000000","updated":"2024-03-26 11:40:10.000000000","submitted":"2024-03-26 11:39:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913738","meta_rev_id":"84956127f24bf058cf6030bcb95b40e904649bea","_number":913738,"virtual_id_number":913738,"owner":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-26 11:39:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"all":[{"value":1,"date":"2024-03-26 08:50:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":2,"date":"2024-03-26 09:57:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"value":1,"date":"2024-03-26 09:57:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}],"reviewers":{"REVIEWER":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}],"CC":[{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-20 08:57:40.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":34680,"name":"Numan Menekşe","email":"numan.menekse@tubitak.gov.tr","username":"numanmenekse"},"state":"CC"},{"updated":"2024-03-20 08:57:40.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"state":"CC"},{"updated":"2024-03-20 08:57:40.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"CC"},{"updated":"2024-03-20 08:57:40.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"},{"updated":"2024-03-20 08:57:40.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2024-03-20 08:57:40.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2024-03-20 10:07:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-20 10:08:09.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"REVIEWER"}],"messages":[{"id":"4e2cdcca634179a2ec9d5802275874858d1c684d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-03-20 08:53:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22a81b5fa7cef2f475d7c5cc2deb3ee6ea30caf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 10:07:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7df4ab76b31643a28ddeb4414dff14e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eecb916d8f4747d4a5f5182b05ff39bb : SUCCESS in 5m 17s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/57f40416399e43c5ac3183a6a40a16f9 : SUCCESS in 10m 30s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/6636beeef36d4354885818ad22d48693 : SUCCESS in 7m 22s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/857c7608ea534a55a79eebd1cedff466 : SUCCESS in 5m 12s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a6c2aaec1ad04658a5e26110b0502084 : SUCCESS in 1h 06m 24s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/384451deaf25460b9ea5ef8c5fd384c6 : SUCCESS in 57m 54s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/56681029408746bb9991e352e573d321 : SUCCESS in 48m 41s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c877c1b74b2242ddb98ce6fa5f841925 : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":1},{"id":"729f69d5f40891130e66ae40409dec3923b2e222","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-03-20 10:08:09.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1d71cd8f294e70d7c224f15b67199cadd5915b9","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-03-22 01:30:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"06f80b0f4fa83501bd5290806934939cc42a449b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-03-22 07:26:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d784c9ca0dcb581298f6968a28b9c9d0fe0e6bf3","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-03-22 07:26:50.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"695cd559f3c0e642d43b6c9b727554ad0db7e911","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 08:39:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ce5f2064b8b465981e69b5d31060831\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5b2ebc524f44c3ba7aa0f9e1afe40e4 : SUCCESS in 3m 53s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/631816637ce74b819face7f2ed591948 : SUCCESS in 5m 48s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/a1d4908a900e4418858b9415e38ead12 : SUCCESS in 4m 17s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/46798b804bbf4398a4138177cba92719 : SUCCESS in 4m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9ce5633d81a745459841ab1fa7fdf581 : SUCCESS in 59m 08s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f1e83bff63ba4b58bfbbd9fc1409056c : SUCCESS in 1h 07m 24s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a680c50169d546f6bdc6664ce9c6b62b : FAILURE in 14m 26s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b9b6a425fc6d455b81c3a4413f57fd0f : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":2},{"id":"4e493075898f10f4e17fc60e2bf97ef859d2f734","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-03-22 08:52:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"531030c07fd750c7f97f4aacb74f58266bff6362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 10:28:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b41d164695b847caa1b75d08ee116d48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce6e13cefaaf4ea8877f81cf51782eb1 : SUCCESS in 4m 49s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/dce26324aa514827a811e967dee4f303 : SUCCESS in 7m 58s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/abd30389f9b84ecd9c3f22ded10f44ef : SUCCESS in 6m 54s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a36f604c04c9487da72aa41004178fc8 : SUCCESS in 6m 41s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/cdd65f6f8b7e4643bc7a654d82698eb2 : SUCCESS in 1h 25m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/6a222aa75cb7472793b19c1fd7fdcfcf : SUCCESS in 46m 56s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/80bbc11e27834badaa357dc964bc8158 : SUCCESS in 49m 54s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7bf4ea9a7e3d4d08aa8ab357493725e2 : SUCCESS in 44m 08s","accounts_in_message":[],"_revision_number":2},{"id":"e5a597605dbeed896e2a30dd8d278fabf998b5c6","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-03-26 01:39:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4edd399e12e1baaaf64f9b8e0d50a20529d650c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-03-26 07:14:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e96470690615c9d22702ed596f41d1ce21bd07d9","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2024-03-26 07:15:18.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4a225a7fd931a0b4d1f493482c555fe8a6eb345b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 08:44:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e31cf80843d843f6ae4cdbf6d26137f7\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/519ef721f0b34ccfbb8d0c459868a897 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23c9e383676d4934b57bddda70d93c2e : SUCCESS in 5m 18s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/215b28bd70da49bbbaafe2dd81bb3a66 : SUCCESS in 9m 18s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b7a0d5087470462081942f6a5d547d83 : SUCCESS in 6m 22s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/99379d8efcc54cdaa4f29f2a535f32c5 : SUCCESS in 6m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/01c6ccb1516341c28199c757a12bc955 : SUCCESS in 1h 23m 36s","accounts_in_message":[],"_revision_number":3},{"id":"8229da581cbc05f0b333814a7cea704961c1c0b8","author":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"date":"2024-03-26 08:50:14.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8e92d803d1265f138ad8a3a5b468798a81e9d3ca","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-03-26 09:57:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"669ab390259a50b390442951f0bfbf876b86a737","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 09:58:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a95bcc6ea985dfd0c5a59d34884ac18aba27085e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 11:39:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0084f3621204624809d862aa3d5bcf9\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9877048ea8014757a0cca1a92b88dff9 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93e62a1a9c9842b8bfe8413df2b5ad64 : SUCCESS in 4m 58s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/393f2bd6fb4f4efdbc0c1c790b833230 : SUCCESS in 11m 08s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/de17917ab95048b190e0f6d0bf0deaae : SUCCESS in 8m 03s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/ac2cd5ba83044befa5cf0ebbf12a8ea6 : SUCCESS in 7m 50s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bc5fa68c573a4af8be34d5f10cc39320 : SUCCESS in 1h 30m 43s","accounts_in_message":[],"_revision_number":3},{"id":"409637f8504febe17f77228e865cb5d2f7619958","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 11:39:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"84956127f24bf058cf6030bcb95b40e904649bea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 11:40:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e156a4cb30e74482a5cc57415aca00cc\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f41efca045849278e70f37e7930e4f9 : SUCCESS in 43s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/163d8db911244a7ca62938be00184b80 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c5fa4cb717d87fe092d6670c5f9ec94e91ab314e","revisions":{"791ba200a8389cd439857f37ed078cdd9a54771f":{"kind":"REWORK","_number":1,"created":"2024-03-20 08:53:32.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/38/913738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/38/913738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/913738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/913738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/913738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/38/913738/1"}}},"commit":{"parents":[{"commit":"e1a9d17f4cb85cbd5887452df1c591c64d726ad1","subject":"fix: fix subnet cidr tip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e1a9d17f4cb85cbd5887452df1c591c64d726ad1"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2024-03-20 08:17:26.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2024-03-20 08:52:56.000000000","tz":0},"subject":"feat: Adding log feature to Nova instance detail page.","message":"feat: Adding log feature to Nova instance detail page.\n\nLog tab has been added to the Nova instance detail interface.\n\nChange-Id: I42005ff85d5c329aaf4a18a9d6181e901d091217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/791ba200a8389cd439857f37ed078cdd9a54771f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/791ba200a8389cd439857f37ed078cdd9a54771f"}]},"branch":"refs/heads/master"},"18ec9fe325648b2a6f70ca99af5022c65284b886":{"kind":"REWORK","_number":2,"created":"2024-03-22 07:26:21.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/38/913738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/38/913738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/913738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/913738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/913738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/38/913738/2"}}},"commit":{"parents":[{"commit":"e1a9d17f4cb85cbd5887452df1c591c64d726ad1","subject":"fix: fix subnet cidr tip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e1a9d17f4cb85cbd5887452df1c591c64d726ad1"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2024-03-20 08:17:26.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2024-03-22 07:25:19.000000000","tz":0},"subject":"feat: Adding log feature to Nova instance detail page.","message":"feat: Adding log feature to Nova instance detail page.\n\nLog tab has been added to the Nova instance detail interface.\n\nChange-Id: I42005ff85d5c329aaf4a18a9d6181e901d091217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/18ec9fe325648b2a6f70ca99af5022c65284b886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/18ec9fe325648b2a6f70ca99af5022c65284b886"}]},"branch":"refs/heads/master"},"c5fa4cb717d87fe092d6670c5f9ec94e91ab314e":{"kind":"REWORK","_number":3,"created":"2024-03-26 07:14:58.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/38/913738/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/38/913738/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/913738/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/913738/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/913738/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/38/913738/3"}}},"commit":{"parents":[{"commit":"e1a9d17f4cb85cbd5887452df1c591c64d726ad1","subject":"fix: fix subnet cidr tip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e1a9d17f4cb85cbd5887452df1c591c64d726ad1"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2024-03-20 08:17:26.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2024-03-26 07:14:01.000000000","tz":0},"subject":"feat: Adding log feature to Nova instance detail page.","message":"feat: Adding log feature to Nova instance detail page.\n\nLog tab has been added to the Nova instance detail interface.\n\nChange-Id: I42005ff85d5c329aaf4a18a9d6181e901d091217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c5fa4cb717d87fe092d6670c5f9ec94e91ab314e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c5fa4cb717d87fe092d6670c5f9ec94e91ab314e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
