)]}'
{"id":"openstack%2Fskyline-console~913872","triplet_id":"openstack%2Fskyline-console~master~I15a0eee4760100d923bc9a7d3619f5caf35cbe27","project":"openstack/skyline-console","branch":"master","topic":"router","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-03-22 03:49:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I15a0eee4760100d923bc9a7d3619f5caf35cbe27","subject":"feat: support disable/enable snat","status":"MERGED","created":"2024-03-21 08:22:43.000000000","updated":"2024-03-22 03:50:58.000000000","submitted":"2024-03-22 03:49:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913872-router","meta_rev_id":"aaca261a91c2f21253ba9fcf89cfeee53c9ac981","_number":913872,"virtual_id_number":913872,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-22 03:49:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-03-22 02:45:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-03-22 02:45:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-21 09:33:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-22 02:45:35.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"56f6ae45285a5db552755377a52b17f9c33e14b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-03-21 08:22:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a4660e9f1af7cd5cb1622258204615af4fb439b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 09:33:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/581e122ac8d94f698162d1d71ebb6cf4\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1545e2424adb4c1fb067a2973bad95d9 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8345a9ea19de426d90a2a6646e2a30dd : SUCCESS in 4m 12s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/8d79661ddf314c9bb017f7a43de81a1d : SUCCESS in 11m 06s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/b226633d95b149f0afb4a8503cf08aa1 : SUCCESS in 7m 36s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/2d915af44fe248b6bdf62d6c70f57a2f : SUCCESS in 6m 46s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b1730f8498264f08a6d617b2c544497a : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8a52613f4a99a586e7e6566b52a4761214304b6d","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-03-22 02:45:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"264d79e5fbb665f9b903041ca28eb6f9eab7b13b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 02:46:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d4cf9ecc22dd2c1db8f57d871e05dfbdd46a638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 03:49:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4990ae14b19d4b64ac4f5c406cc774e4\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e67e860151c84ddc8d98810e190a3c93 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49932fc3b78f45c481ad8b7ad9e0cd6c : SUCCESS in 4m 14s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d4dc9e70210e4ee08d40faabe3340b08 : SUCCESS in 10m 45s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e08fdd8d55e14eb9979b9270bb3f3ca0 : SUCCESS in 5m 21s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/4f8699ee6d994fa9a2af5359aa7837ae : SUCCESS in 6m 25s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/313b3d4009e943bf9a3f1241ed76b4d0 : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":1},{"id":"eec33318aa16c2972e68c9a4ed8e256166ca22e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 03:49:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aaca261a91c2f21253ba9fcf89cfeee53c9ac981","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 03:50:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a6d646da9b9431d8d19e3bafc2b7837\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/972bb18bcf8f421f8ec5bce8a9705c6d : SUCCESS in 45s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94ddd4ffb723401e95101dc72c5726a1 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d02497a15d1bc35a0532b524a3a4a9f86258f050","revisions":{"d02497a15d1bc35a0532b524a3a4a9f86258f050":{"kind":"REWORK","_number":1,"created":"2024-03-21 08:22:43.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/72/913872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/72/913872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/913872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/913872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/913872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/72/913872/1"}}},"commit":{"parents":[{"commit":"e1a9d17f4cb85cbd5887452df1c591c64d726ad1","subject":"fix: fix subnet cidr tip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e1a9d17f4cb85cbd5887452df1c591c64d726ad1"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-21 08:22:14.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-21 08:22:14.000000000","tz":480},"subject":"feat: support disable/enable snat","message":"feat: support disable/enable snat\n\n1. support disable snat when the router\u0027s enable_snat is true\n2. support enable snat when the router\u0027s enable_snat is false\n\nChange-Id: I15a0eee4760100d923bc9a7d3619f5caf35cbe27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d02497a15d1bc35a0532b524a3a4a9f86258f050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d02497a15d1bc35a0532b524a3a4a9f86258f050"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
