)]}'
{"id":"openstack%2Fskyline-console~914564","triplet_id":"openstack%2Fskyline-console~master~Ife201c5f04766f1a304b1bb1b562e643ad5ed0cb","project":"openstack/skyline-console","branch":"master","topic":"monitor","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-03-28 07:19:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ife201c5f04766f1a304b1bb1b562e643ad5ed0cb","subject":"fix: physical node page display error","status":"MERGED","created":"2024-03-28 02:42:20.000000000","updated":"2024-03-28 07:20:47.000000000","submitted":"2024-03-28 07:19:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914564-monitor","meta_rev_id":"8478c486f74fb6791246870a14846121add67d67","_number":914564,"virtual_id_number":914564,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-28 07:19:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-03-28 05:50:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-03-28 05:50:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-28 04:07:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-28 05:50:04.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"8a3ea2e45936aa7bc7418e8bbf85f52adf4381e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-03-28 02:42:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4250e06544ff47f141435532bad9fcfdd236eb3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 04:07:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9b51a895b594fb2820a714bfa7177b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25fd55b2d331430ab3419efbf8703ba2 : SUCCESS in 5m 51s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/4efce0d94b27487d968c35e4b9b3d3ce : SUCCESS in 9m 36s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/086a2ad6beec4000a834ea90ada47b41 : SUCCESS in 6m 51s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/82c8a74142734dbd8f9f6a2ebf340b71 : SUCCESS in 4m 54s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6ed0096850144d22aa66fc8704873529 : SUCCESS in 1h 09m 07s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2a6aeb7e27aa45a4843a1b1ed0975430 : SUCCESS in 1h 06m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/45507806436d4a1b9be5742a953162e3 : SUCCESS in 1h 01m 31s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/493ae95592a445349adfa9d066f16607 : SUCCESS in 48m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2fb59557cc889a799db4faee43963ae5dcbafff9","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-03-28 05:50:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f295d8e757c5120186514d34230267b95235e3a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 05:50:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bae2c36831bad06a47a170cbb636bd59f069fad2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 07:19:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fe7bd7e75aa495ba5cb2082872b9232\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1756ced686a45c2a79638f423ef93ef : SUCCESS in 5m 05s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/2f685c7cff6d46c9810ae570125ec6f0 : SUCCESS in 6m 29s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/3682efa6bde34f059ec640a6e7bd90d6 : SUCCESS in 7m 37s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/c8f4ceb4c5f746e6926e226a29717825 : SUCCESS in 7m 59s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c1ce40b7c7964e5b9d8b5953d4f68413 : SUCCESS in 1h 24m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5d4723e269d845208bb0bbd25c33f4ec : SUCCESS in 56m 16s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3aa1c080c4d841a8a26beb418121fbba : SUCCESS in 44m 09s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c1f66ed6b3ed421693fb749942c2e8de : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b16815aac236bea35fa892fa1ecf13cda1cc72be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 07:19:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8478c486f74fb6791246870a14846121add67d67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 07:20:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2f7fc6862f342668bcb01c0a7ba4bf7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfb5736180e344ae938c4ab81fce503b : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46f3668f8ed53c82191cb1b9e5eea844bf4789e0","revisions":{"46f3668f8ed53c82191cb1b9e5eea844bf4789e0":{"kind":"REWORK","_number":1,"created":"2024-03-28 02:42:20.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/64/914564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/64/914564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/914564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/914564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/64/914564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/64/914564/1"}}},"commit":{"parents":[{"commit":"8613ef39f30343dbb2f030e1c73fa76a0eebc47e","subject":"fix: fix create/delete encryption for volume type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8613ef39f30343dbb2f030e1c73fa76a0eebc47e"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-28 02:41:57.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-28 02:41:57.000000000","tz":480},"subject":"fix: physical node page display error","message":"fix: physical node page display error\n\nFix display error when node information cannot be obtained\n\nChange-Id: Ife201c5f04766f1a304b1bb1b562e643ad5ed0cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/46f3668f8ed53c82191cb1b9e5eea844bf4789e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/46f3668f8ed53c82191cb1b9e5eea844bf4789e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
