)]}'
{"id":"openstack%2Fskyline-console~914569","triplet_id":"openstack%2Fskyline-console~master~Ia0fbe481adc9cdfe9b4179c611283d16983e4dfd","project":"openstack/skyline-console","branch":"master","topic":"i18n","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-03-28 03:52:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0fbe481adc9cdfe9b4179c611283d16983e4dfd","subject":"feat: update i18n for instance console log","status":"MERGED","created":"2024-03-28 02:59:59.000000000","updated":"2024-03-28 03:54:12.000000000","submitted":"2024-03-28 03:52:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914569-i18n","meta_rev_id":"79efe436bb046657824b59acde1760704d3a0835","_number":914569,"virtual_id_number":914569,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-28 03:52:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-03-28 03:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-03-28 03:21:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-28 03:19:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-28 03:21:51.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"b355387dedba921f0e7c1298656826df60f16cbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-03-28 02:59:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adca99b104b956b2be6b8e1735a7cf5458544767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 03:19:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fec55878f60342c69c35cd672ea3ebd8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b828960fa1c4546a341b9bc05c82afd : SUCCESS in 6m 03s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/3614a66d4ca84a6681dd9b6c007d5017 : SUCCESS in 10m 20s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/f724dd4c751347bcad6f4e3fe235307b : SUCCESS in 6m 32s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/98cd09bc6748402da75f5e22ce252b61 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c7bb22c7e3a8e03905b0b01c5905f80b8ac2c465","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-03-28 03:21:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"562c8957165decbcf00481838cfdb23ef265c219","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 03:22:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f77ea74046f52d10443c1749e4d3bf4a1bdfc8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 03:52:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3feb28df77b447fcb4f07c21d5060878\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b250c5d18670496b82dd5c5aabe0789f : SUCCESS in 4m 39s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/0ddbc62fae1d47de82b294575d2f63e1 : SUCCESS in 8m 31s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/20857b44d5ab4a7daa305c684976de9b : SUCCESS in 7m 18s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/b69ee33594554106ac3f9099665eee74 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"172eb23575b4195b2e4156f53f25261b15db7cf6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 03:52:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"79efe436bb046657824b59acde1760704d3a0835","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 03:54:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f54af5fb4eb44b9aa6d0830d876a2a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f80aa50e3054037a25586b1f54b3d10 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7639237e5b2dd850868c89c2405e7f0828044af6","revisions":{"7639237e5b2dd850868c89c2405e7f0828044af6":{"kind":"REWORK","_number":1,"created":"2024-03-28 02:59:59.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/69/914569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/69/914569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/914569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/914569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/914569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/69/914569/1"}}},"commit":{"parents":[{"commit":"8613ef39f30343dbb2f030e1c73fa76a0eebc47e","subject":"fix: fix create/delete encryption for volume type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8613ef39f30343dbb2f030e1c73fa76a0eebc47e"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-28 02:59:40.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-28 02:59:40.000000000","tz":480},"subject":"feat: update i18n for instance console log","message":"feat: update i18n for instance console log\n\nupdate i18n for instance console log pae\n\nChange-Id: Ia0fbe481adc9cdfe9b4179c611283d16983e4dfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7639237e5b2dd850868c89c2405e7f0828044af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7639237e5b2dd850868c89c2405e7f0828044af6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
