)]}'
{"id":"openstack%2Fskyline-console~914581","triplet_id":"openstack%2Fskyline-console~stable%2F2024.1~I28e00000353e4929c13c74264582dc14d6e726ed","project":"openstack/skyline-console","branch":"stable/2024.1","topic":"create-instance-stable/2024.1","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-03-28 10:20:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I28e00000353e4929c13c74264582dc14d6e726ed","subject":"fix: fix subnet cidr tip","status":"MERGED","created":"2024-03-28 06:44:47.000000000","updated":"2024-03-28 10:22:00.000000000","submitted":"2024-03-28 10:20:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914581-create-instance-stable/2024.1","cherry_pick_of_change":913615,"cherry_pick_of_patch_set":2,"meta_rev_id":"340d86b3ea12c453ea05778e747b1a2e7010bf99","_number":914581,"virtual_id_number":914581,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-28 10:20:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-03-28 08:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-03-28 08:20:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-28 08:03:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-28 08:20:32.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"a7e00de621b1aa8201ed57492ab845e7174c2205","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-03-28 06:44:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"34d66cd1ecc0ece35c7b842817e05c5d10f79d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 08:03:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06ace3859bba4111884f55d68ebec592\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95aa22abad884b86934bfe13e80f6e32 : SUCCESS in 4m 22s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/aa8c806a9ce74d1a8757f2cabab8de31 : SUCCESS in 5m 57s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/32c3442949764f209b4b0dd9d79e8c77 : SUCCESS in 6m 49s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/a6570417a4ad452fbecd1a2ac771fb18 : SUCCESS in 6m 46s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d004e65185d744d29694bf5ca7240c46 : SUCCESS in 58m 19s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d1f945750dcc4ac9850fe9daac8d0af6 : SUCCESS in 1h 07m 20s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1fcbf4c1489e4201a213b3f1c18eaaff : SUCCESS in 1h 01m 22s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/53b3750eaea94d7a980e64956c815d83 : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":1},{"id":"53a2c260814507945527dfd95a933487d8de650b","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-03-28 08:20:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2b825266d39dfb588763f7cb2308a2df5e23fa81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 08:21:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c70e59a34c1093ed80bf73361d25a80952dcfce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 10:20:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/059155082f89426f99aebdd38a55167f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0213bdaf6a0741af97590aa9c86de55d : SUCCESS in 4m 21s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/c3d6e59b1d9f48fc814bcc90299bab15 : SUCCESS in 9m 50s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/ab18e2dce45e4447818958c6c3f0ffd1 : SUCCESS in 7m 02s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/6e34313206e74f81bfee5c2c247a60e7 : SUCCESS in 5m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4a1d29d9bfec429d91ee284fca92dc30 : SUCCESS in 1h 24m 27s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3675d52954c14436a9641b96791deea1 : SUCCESS in 1h 03m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/05c38f1fb6594980ac464465d3ee629d : SUCCESS in 51m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/164776a92a2d40b4931d58f41fb77873 : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a5fa67bd29501d87ea2e9ffd748a4575d220880d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 10:20:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"340d86b3ea12c453ea05778e747b1a2e7010bf99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 10:22:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/214cf5e154a9429e8b9ce1d3460d1a0d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de8c15ca52a485a8f5bac2a8f3565ad : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ddeff6447e87d1d0a30e7fa3c09d2876ec5a023","revisions":{"7ddeff6447e87d1d0a30e7fa3c09d2876ec5a023":{"kind":"REWORK","_number":1,"created":"2024-03-28 06:44:47.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/81/914581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/81/914581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/81/914581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/81/914581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/81/914581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/81/914581/1"}}},"commit":{"parents":[{"commit":"f40bde40a0e898936976a194b8829de6a2eccb4d","subject":"fix: fix create security group rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f40bde40a0e898936976a194b8829de6a2eccb4d"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-03-19 04:03:26.000000000","tz":480},"committer":{"name":"Jingwei.Zhang","email":"zhang.jingwei@99cloud.net","date":"2024-03-28 06:44:47.000000000","tz":0},"subject":"fix: fix subnet cidr tip","message":"fix: fix subnet cidr tip\n\n1. fix subent cidr tip when select network\n2. update subnet selector style\n\nChange-Id: I28e00000353e4929c13c74264582dc14d6e726ed\n(cherry picked from commit e1a9d17f4cb85cbd5887452df1c591c64d726ad1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ddeff6447e87d1d0a30e7fa3c09d2876ec5a023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ddeff6447e87d1d0a30e7fa3c09d2876ec5a023"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
