)]}'
{"id":"openstack%2Fskyline-console~914801","triplet_id":"openstack%2Fskyline-console~master~I4e167cfbb3e29773ef6bdd56b84e5c883bf53920","project":"openstack/skyline-console","branch":"master","topic":"volume-type","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-04-01 09:09:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e167cfbb3e29773ef6bdd56b84e5c883bf53920","subject":"feat: update qos spec column for volume type","status":"MERGED","created":"2024-04-01 06:48:34.000000000","updated":"2024-04-01 09:11:00.000000000","submitted":"2024-04-01 09:09:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914801-volume-type","meta_rev_id":"9189b9ee53d020618d19609149eff2290b1773f6","_number":914801,"virtual_id_number":914801,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-01 09:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-04-01 08:01:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-04-01 08:01:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-01 07:51:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-01 08:01:58.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"c7109d445bf2ec0bb6c2888e1ee1c969b84f21df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-04-01 06:48:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31d877ecdbbcb0b706549b0453c7c7238454209d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 07:51:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbe63f135cc54e41bf76496dc0299f47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/390ceaaecbc84176919a15e5c25e544f : SUCCESS in 4m 32s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d9f1fa76e37b41d19a2866ad8895da14 : SUCCESS in 8m 58s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/9b83e17d4707436a88c20baf06b0ae2a : SUCCESS in 6m 35s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/3147f435a4cb4f7ab1073308b445f924 : SUCCESS in 4m 09s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a11084a9804d4cfabdee6f698283dba5 : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":1},{"id":"764b0c0d57dfd49e767ca886191b2dae96769d1a","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-04-01 08:01:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f60f3c9812166816aff349de380aeadeaeef345e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 08:02:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88d03a0d8c894ce15a5523dee79faa3f70121f0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 09:09:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b92e7d2e62046b0bbb95170b765e17a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f235b5e9b6854b0fadbf2b8c49f6029c : SUCCESS in 5m 16s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/18bca6612322482abfeab4b13be8030e : SUCCESS in 10m 16s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/47bdc56cfc75441297cbd620d4b7f297 : SUCCESS in 5m 13s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/6721303ee90247b99a49bf4b0093a9c8 : SUCCESS in 7m 00s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b4b17d7e29ad4cd993c86d38c3458690 : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b7c97f75359f90624315de592235389c489b63d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 09:09:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9189b9ee53d020618d19609149eff2290b1773f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 09:11:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/522cfdbbe309472eaa92855d587240dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fffe80de7907462bb5d42679344203dd : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae3cfafb5c5e3e195f949e7228c86cd5ef26f1f0","revisions":{"ae3cfafb5c5e3e195f949e7228c86cd5ef26f1f0":{"kind":"REWORK","_number":1,"created":"2024-04-01 06:48:34.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/01/914801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/01/914801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/01/914801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/01/914801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/01/914801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/01/914801/1"}}},"commit":{"parents":[{"commit":"4cc407f740bf120590b5374315cf3d48e935a15e","subject":"feat: update file system space info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4cc407f740bf120590b5374315cf3d48e935a15e"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-01 06:47:32.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-01 06:47:32.000000000","tz":480},"subject":"feat: update qos spec column for volume type","message":"feat: update qos spec column for volume type\n\n1. update qos spec column for the volume typs list page, then the download file can contain the qos spec id.\n2. update the label for the qos spec id search.\n\nChange-Id: I4e167cfbb3e29773ef6bdd56b84e5c883bf53920\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ae3cfafb5c5e3e195f949e7228c86cd5ef26f1f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ae3cfafb5c5e3e195f949e7228c86cd5ef26f1f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
