)]}'
{"id":"openstack%2Fskyline-console~914803","triplet_id":"openstack%2Fskyline-console~master~I744d1b87a9e9545968cac12bae4f8cf257f3f526","project":"openstack/skyline-console","branch":"master","topic":"base-components","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-04-01 10:18:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I744d1b87a9e9545968cac12bae4f8cf257f3f526","subject":"feat: limit the height of prompts","status":"MERGED","created":"2024-04-01 07:17:52.000000000","updated":"2024-04-01 10:19:24.000000000","submitted":"2024-04-01 10:18:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914803-base-components","meta_rev_id":"4918c336eb890f6fb17b04c7db2f23957b075c15","_number":914803,"virtual_id_number":914803,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-01 10:18:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-04-01 09:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-04-01 09:06:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-01 09:03:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-01 09:06:36.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"2cbebf22f710347abd64feaee255974942ad1cf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-04-01 07:17:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a529d305e06d4fb7902fb5a2209cff8f6689984a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 09:03:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd10c20fd23940c49c278ab986561bad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e6a24396d8d4d118ad04dcf06012bd3 : SUCCESS in 3m 46s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/d19295dc8d7846b5b45baea5c97121d7 : SUCCESS in 10m 26s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/9bb22f9e63384cd7b93171b1aa04be05 : SUCCESS in 6m 29s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/277fa47d89db4a70a710eb15783caa7e : SUCCESS in 6m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b674ac6d80ee4e4287ba5b6456191bc3 : SUCCESS in 1h 27m 24s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9c1acc1c40fc4936b3e595fa7f5c086e : SUCCESS in 1h 04m 10s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b7d9320d13544ff19e34da4cd20e02fd : SUCCESS in 38m 36s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2531e325e16345148cc06e807c58fb4a : SUCCESS in 1h 00m 33s","accounts_in_message":[],"_revision_number":1},{"id":"18323bf70cc7ab46c02f4ff6c153cd94bba45d34","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-04-01 09:06:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ec926a6e11a37cebaa93bb7ec75d979f890623b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 09:07:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d77a39370d9052db93484d09cb8fe0947babb95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 10:18:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9096ae1252d64b1295c025e519e1e657\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/716c5f53302c47a7b6250709432fbf3c : SUCCESS in 4m 11s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/586e9bf5bfcf4b13a7fadcc25c57dc72 : SUCCESS in 6m 12s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/8c41364119bc40768950434dd8bdf75b : SUCCESS in 4m 28s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/dd794c218882492f833f42ae671b5d1f : SUCCESS in 4m 31s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b49a42bf2bab4a979b37c1b26428dfda : SUCCESS in 1h 01m 25s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d8dcfe9c54ca4305ad0ec11681fd6bc6 : SUCCESS in 1h 06m 01s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c3312bb711f041abbfcbc43c1d5fa65f : SUCCESS in 41m 42s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/30675a9176894d35a4755f9ab932dbea : SUCCESS in 38m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4ab466e6d2584faa1cbcfcd40afe52546c5f1556","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 10:18:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4918c336eb890f6fb17b04c7db2f23957b075c15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 10:19:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a1952d0858f40d6a09c4db1260e0f92\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea796ad2320343a0968ba823f82dfc50 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e825d510befe85189dca5c0942b1716a284a4feb","revisions":{"e825d510befe85189dca5c0942b1716a284a4feb":{"kind":"REWORK","_number":1,"created":"2024-04-01 07:17:52.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/03/914803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/914803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/914803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/914803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/914803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/914803/1"}}},"commit":{"parents":[{"commit":"4cc407f740bf120590b5374315cf3d48e935a15e","subject":"feat: update file system space info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4cc407f740bf120590b5374315cf3d48e935a15e"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-01 07:15:07.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-01 07:15:07.000000000","tz":480},"subject":"feat: limit the height of prompts","message":"feat: limit the height of prompts\n\nIn order to display table data well in the list page, limit the height of prompts\n\nChange-Id: I744d1b87a9e9545968cac12bae4f8cf257f3f526\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e825d510befe85189dca5c0942b1716a284a4feb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e825d510befe85189dca5c0942b1716a284a4feb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
