)]}'
{"id":"openstack%2Fskyline-console~914856","triplet_id":"openstack%2Fskyline-console~master~I2479007e4c564e059aedd032e230bb7930191c3b","project":"openstack/skyline-console","branch":"master","topic":"base-components","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2024-04-03 08:12:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2479007e4c564e059aedd032e230bb7930191c3b","subject":"refactor: refactor NameInput component","status":"MERGED","created":"2024-04-02 09:22:02.000000000","updated":"2024-04-03 08:13:52.000000000","submitted":"2024-04-03 08:12:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914856-base-components","meta_rev_id":"36d8e5bdca397e115d2358a09bd0c3487dcfc331","_number":914856,"virtual_id_number":914856,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-03 08:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":2,"date":"2024-04-03 06:55:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":1,"date":"2024-04-03 06:55:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-02 10:58:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-03 06:55:11.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"}],"messages":[{"id":"32211aa2d3dc29375eef7aec4b2b7a395a070d24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2024-04-02 09:22:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"719b581cce6200d0aa8323281a8102f872eecee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 10:58:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82d187161f5840ca87fbcd3c688b952b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01462c9da6294dfa9fb534e4c1905ba2 : SUCCESS in 5m 21s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/00f354b5f80a4a2b8e35764268451890 : SUCCESS in 11m 10s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/8d153ea0391a4bdd9754c1c771d0f510 : SUCCESS in 7m 50s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/237603f3d8fd4404b04eea431b68cec3 : SUCCESS in 8m 17s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/70a8019ab53b4d8f97a567a11a825d13 : SUCCESS in 1h 19m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ab5331e922774702b1eb86343a5bdd80 : SUCCESS in 48m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9f6ed4b3956e450886cb328dca975d3d : SUCCESS in 45m 25s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2e8de935454b4f2fa6a43475eabf50a0 : SUCCESS in 1h 01m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bba27ba764a6bebbdcc9e9dec6de2bdad2c0a60d","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2024-04-03 06:55:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4c1ac68d90d47311813993053072aa8e1057cd5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 06:55:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a30d97f10f920a07bb05875a9029a61c394a1d8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 08:12:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f9a679a59d84f5ba18b0dc8c3df3ae9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0bdfdfd68d245638d420e2bfe44d3c2 : SUCCESS in 4m 39s\n- skyline-nodejs14-run-lint-src https://zuul.opendev.org/t/openstack/build/61188bc96c404a549944f420b4d2c167 : SUCCESS in 6m 51s\n- skyline-nodejs14-run-lint-test https://zuul.opendev.org/t/openstack/build/e31dc298c398442580cf1bf42c4a3812 : SUCCESS in 5m 06s\n- skyline-nodejs14-run-unit-test https://zuul.opendev.org/t/openstack/build/6e0109a26a0f4ef793a4cd6d5de20792 : SUCCESS in 7m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f87c692f4b0148dba477401ceef376dc : SUCCESS in 1h 00m 30s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/db931d32e4784f5889e924ec417572ca : SUCCESS in 1h 08m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/09433f189cb04b15ab8e790e1399e1a3 : SUCCESS in 59m 08s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1adb7738fae74331a3e2ac321a7b020f : SUCCESS in 38m 09s","accounts_in_message":[],"_revision_number":1},{"id":"63fff2192167e7997f68c36690720542380b9f6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 08:12:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"36d8e5bdca397e115d2358a09bd0c3487dcfc331","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 08:13:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e78bf9e7f5764f6bbf077822877c4588\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1944edefa9e474f8146bb6d6c810677 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9215b712e31e068404d30f0467b473eb384363fd","revisions":{"9215b712e31e068404d30f0467b473eb384363fd":{"kind":"REWORK","_number":1,"created":"2024-04-02 09:22:02.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/56/914856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/56/914856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/914856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/914856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/914856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/56/914856/1"}}},"commit":{"parents":[{"commit":"36483416f23f4b2aa915efc91dc1d5a1af289a19","subject":"Merge \"feat: limit the height of prompts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/36483416f23f4b2aa915efc91dc1d5a1af289a19"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-02 09:21:46.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@99cloud.net","date":"2024-04-02 09:21:46.000000000","tz":480},"subject":"refactor: refactor NameInput component","message":"refactor: refactor NameInput component\n\nrefactor NameInput component which is used in form\n\nChange-Id: I2479007e4c564e059aedd032e230bb7930191c3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9215b712e31e068404d30f0467b473eb384363fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9215b712e31e068404d30f0467b473eb384363fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
